)]}'
{"id":"openstack%2Frequirements~978295","triplet_id":"openstack%2Frequirements~master~Ifdce39f736a5a61586e304c8978d4d159ab0bbd2","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-03 07:14:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdce39f736a5a61586e304c8978d4d159ab0bbd2","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-02-28 08:15:58.000000000","updated":"2026-04-03 07:14:49.000000000","submitted":"2026-04-03 07:14:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978295","meta_rev_id":"dabc4daf67c3a602c8e2d2fdedd6769aa6fdfd34","_number":978295,"virtual_id_number":978295,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-03 07:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-02 15:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-03 05:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-03 05:30:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 08:19:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-02-28 09:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-03 05:30:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b61fe453797d81cb335d9c762033a1c3516050fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-28 08:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d6c2c287adeaca916d9b238a7bbe8a1e96054d1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-28 08:19:11.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0af1fcebc173e330b5aee3d9c9daaa64acf00849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 09:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e31389e054684fd19c5bb8fff084219c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f80b58289b4c0a9072faecdc69fe9b : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d450184c6102461e9a79b8c761dbffe7 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37960344568a4f2ab0799e012fb48b2a : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89adb77a062142b49940970a086f2236 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbdfbe2bd48b433b8a2eb1c0f809798c : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3347288bde404883b5ee23343bc47aab : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b2e26bc941a40cd9b42e4ff4797fc72 : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d954b12798fb4f0b88c070a2a6281a67 : SUCCESS in 4m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/54cfcc35322e4aa89d00c359df8459c9 : FAILURE in 2m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9e7b54530074d31b931d62f28fd2fe7 : FAILURE in 3m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f3b83e976e7c41b9ab4c0f993c2b040f : FAILURE in 2m 34s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/905d58ed04b74af3a6f50b8cf78c1c7c : SUCCESS in 4m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/82013223bb054f0abbc6cafc884a715b : FAILURE in 6m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d57194e4b03c427494ec53610eea6654 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b24689657d4b4dc7bf54ab817e95d0bb : SUCCESS in 10m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e230c6e01caa4f58a51ab3af20fc4628 : SUCCESS in 3m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6044054a1ef148afa03ed4bdcc35ae1c : SUCCESS in 5m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e24164b0e9c04ef68dacb2691ba68381 : SUCCESS in 5m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c4567bf226bf413fad75f005c74258a8 : SUCCESS in 15m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c920ef55ca7445189e0a6118af3d9ea5 : SUCCESS in 4m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5aa21afb21a341fe927683bdc2931bdf : FAILURE in 4m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bf3349bcbf87423b9c6fd29aacd2f7dc : SUCCESS in 10m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f9827231b4594ffda99559d96be49d0f : SUCCESS in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f0d5a1ae43254a6f8958b1a68b481377 : FAILURE in 12m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bd82a1eaced14651baab9682dda1ea18 : SUCCESS in 5m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1356093c2d4147c78cbbd3f556b2e20e : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dafa4e769a1340d3b1a70c5c8fb6efb7 : SUCCESS in 9m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fef494581915410c80fbe0c0da800b29 : SUCCESS in 35m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8536ea4b4e374abe8fa9a0834880e673 : SUCCESS in 8m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b85b45ce96ff4768a8131c8b34c9fd95 : SUCCESS in 13m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8d9eb35256b94b7bb8b555506c3c4b87 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e90ae1199ec4b64bbcc02f6767a0da8 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/778ab88d903f48e690f20d9e4a63b752 : SUCCESS in 27m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8c1539d3220e461098eaab6df7597b18 : SUCCESS in 4m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/559cc71d16b2411885fd9556017a721d : SUCCESS in 2m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af6d6777b32f45c9aace783d9b8485f6 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/523bbaf80a8d49c8b1d5dd242beae7a0 : SUCCESS in 6m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/842080c483fa4ca3a3d19c1ddc64bf73 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05c1463e52a243e4a056c39c7f3b4b37 : SUCCESS in 38m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52142ca1e9114ac9b56fb1b71f6316f3 : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"61771f203e7e0b5d15c0a54489783e7589de7d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-07 08:16:32.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"36f186a02a8ec98890195835e5788e22fd70322d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 10:05:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b72d6e2ab2f4f299348baf473f84210\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99eaa7cbf4524a818292fd5d0e0e702f : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36b120f5594429d9120d02c284d52ee : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cee838a01a9a4f818413c978e1f506e3 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3a2ddcfd3ef4e988be30d9c02a891ff : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83ffd0541d7b4fa2bdfbdaae4fe6bad7 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1966a1a958f4183a72be3b59378d930 : SUCCESS in 1m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e68549df56b847cb848d1f247717d9b7 : SUCCESS in 2m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dfd3049583c84e97af21e98396d1f0d5 : SUCCESS in 4m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/01a15d4cb5c54c2182fad093fcc883ea : FAILURE in 3m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/52a8908adc044223a809bd7382e8fed4 : FAILURE in 4m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b8b2fa154cc14bcf81492a9caef82c90 : FAILURE in 4m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/26cdf668130249da85848702c4b1e9ba : SUCCESS in 2m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d41424fdeb434b71bde0e03c9de9966e : SUCCESS in 5m 44s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/aa863047e79f4e0c913894feb5c437a4 : SUCCESS in 5m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/405cc477e0304abb9a006dc58b05b964 : FAILURE in 10m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8874ba8c3ec14201b00c53d4482277d3 : SUCCESS in 6m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7bb086524e9d4da7a1a77c654e50ae29 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bf232011d978456391f071da66f69395 : SUCCESS in 6m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/47dc8f13d23a42d19600d8b2961883d9 : SUCCESS in 18m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c633883b86a4129bfdae7c2ad9484c8 : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e80ecda3b1234805b07fb4b2b8f9674f : FAILURE in 4m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/81a665f8192d4a288c0a8bb59c3d59cc : SUCCESS in 13m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a0c226af22b44ebcb9472c877b47082f : SUCCESS in 2m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/059c2d8f2e304bb3b98776b855e58b3c : SUCCESS in 9m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7eda43ac6c024191b699666bddfb4326 : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/06c5c510b0a249059362413518fe988a : SUCCESS in 6m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/81699875f4d94a4c99a1a8fe485776ee : SUCCESS in 10m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/acfa810648704bf8b9905ea223c43cca : SUCCESS in 24m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a409eb7da313452d890f62464cf69cd2 : SUCCESS in 8m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/697c9f4ccc3e41408cb37a0e0a5f1dec : SUCCESS in 9m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/71f52ee24f644c11ac24b679e6a708c6 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3da2229af5f64681a8f038e0ffe86f14 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/054f397f50c244e6b80dbdd3a41123d0 : SUCCESS in 26m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d5c9f1bcb3a24aff975d1f8db9933674 : SUCCESS in 6m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/db5e562851994a2c93afc6dbe1eb709e : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c765a4006cd54c0889a1969e9d71d54d : SUCCESS in 6m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cc809430ca984b559066bf33f7263179 : SUCCESS in 8m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/65f34eab3c0542ec9d98179151621c74 : SUCCESS in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfbc5c62d80c461d86211fe64b4d49f4 : SUCCESS in 1h 05m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88e358b5ae234ab3bb8c9fbba0d9bb0a : SUCCESS in 1h 48m 16s","accounts_in_message":[],"_revision_number":2},{"id":"df11f83a1684b24693a89b91174e1e0f8e140687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-14 08:09:26.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e5915161636ccc5f814f275b789ac1d575246b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 10:04:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a2117ada9c6460d8a7a9c40a76cb017\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3151e5f5c70e4bd1ac5fc0dfc2d0a1d7 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc357e4cf034bcf97eefa1b23f3eb1c : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5340d5745c5848f9b8df49d7b89278bd : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba1fded6158941498a93a98a4844a6da : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e3cb57fcd674f5380bcb44b7a88b3ae : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c65f3f7c4ce45c18540ac126835ad59 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d75ab7006fd48489f5c62d60c2cb071 : SUCCESS in 3m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f77eb2bc78db49939582a648381b2ca2 : SUCCESS in 3m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3e944f9007e54cdbbb3b09072ca58b30 : FAILURE in 2m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b8d7462b3d4a468e81adfaa648f107dd : FAILURE in 3m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/05d84a66fda44d2da2c22f4d57ec707d : FAILURE in 4m 12s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9b6e9a62100e457ab63a3b2a767696eb : SUCCESS in 5m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/608d65c222e14d47a55ad13ef47c7a79 : SUCCESS in 5m 16s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a3b67a094e0b491d93d248a1398b48ba : SUCCESS in 4m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e1da96e7b0a741e4a04658c3aafade5a : SUCCESS in 7m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b772e3deb94a40d38187d8d1ebc00b08 : SUCCESS in 6m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/98e1c5cacbef4041804541d3bdf21d8d : SUCCESS in 4m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7d5b7a093a584d528678cdad1500a698 : SUCCESS in 8m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7e641a5e7a1c4a248b4b68fffa1451b1 : SUCCESS in 17m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f120419ed3704dc9bbe971c733609fa6 : SUCCESS in 4m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b1b1eeebf6f34626a5fe0f8df7532352 : FAILURE in 4m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ed02a00c9ccb4fa1b9f0d565672ecad1 : SUCCESS in 12m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a4824b07f4954fbda529c35ead23d1f1 : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/36abb88e0f014ac3b48bdecc6d4cbb8f : SUCCESS in 9m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0f2d81830b584586bc2f3b8741c29cba : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3f5341a2539641b5970eac8bbdad815a : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1b746ebc9c0c40a5bc1357aa6ff848b2 : SUCCESS in 11m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a3aa53fefcab4d3da0d8c3a5405a8208 : SUCCESS in 31m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/632455b879d04e289afcf00440483924 : SUCCESS in 9m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f55b1c99e90940ecba509d22be80128b : SUCCESS in 13m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eadc174ed9764a2ca66e6255e1e6ede9 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c579f83e5ec1401db07c3abae1ecb5cb : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3da24a2601274e42b2d04bb945fbad7a : SUCCESS in 29m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c10d2250415b43b0ae8fb0f7a5c81147 : SUCCESS in 7m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b1800054b69848f48013212e6eed3ef5 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81299f9e704d465184cbf8f79fbab77e : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/21b499c8e1d04b809743c25d06c43c35 : SUCCESS in 7m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/973f76c8b60a4678a74d659365cf50bd : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e926f225c45843c0835bd1dec3f131ec : SUCCESS in 1h 08m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcd8f0fa8867497e8b35e648b826cd94 : SUCCESS in 1h 50m 32s","accounts_in_message":[],"_revision_number":3},{"id":"fce62d812bd36e5479c1de2e73728edd5040355a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-21 08:19:57.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"90ffb0af79dec0baaebb187b02ee398ff59523e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 09:19:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69c57f72c13c4fcc8be4db0d038ffa7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347e5834a8bb4143aaa1c6101395c1a4 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abfd71a421649b48177882a0504fe22 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dbc4552720f4d739677064c42d1c822 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a9a923a9d12449e9fdefa21e8bb9fa7 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/172c51eab1f64f1eb5fc871d8fb491d4 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38312f7a51e94c84a987c7f06c0b53eb : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3d5209803334e6a83966f75289a5ac1 : SUCCESS in 4m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7b712866bc2141baae98cebc0aca3d9d : SUCCESS in 4m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/93cacd4d7c6f45ecb418022cb197d399 : FAILURE in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6901850ada2d4e7f8fc64ffeb91b1e22 : FAILURE in 5m 00s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f753e208f868485ab98585d5f8915085 : FAILURE in 6m 41s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9242fdaf68d44428918bcdc360693ae9 : SUCCESS in 6m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7da8a55765a14a47bc598633e73d98a8 : SUCCESS in 4m 31s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d7257169c3374512b34a82e797edda2e : SUCCESS in 3m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7e3a16ad6fc24a86ba27ae4eab4dd126 : SUCCESS in 11m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3b291eda952d4f049ed2639e19d25eba : SUCCESS in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1196780252234a53a97e8a92981396d3 : SUCCESS in 5m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2868699b0105415189880ded5c86b6c7 : SUCCESS in 4m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ebe451334a84d5bac286e49eea7ea0a : SUCCESS in 19m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/821d7b1112124210a0f120827310f1e9 : SUCCESS in 5m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c3ef2541e1084948978b4b2a9896bafc : FAILURE in 4m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/95d56a65cc11479ca9b6d949fb0c4ae5 : SUCCESS in 9m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d35bde8cdb024b5f903a83d71ea2437f : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ff6fa6fe727449a1be6e5a51db78cf7d : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/88ab8f2e302f40988b5eb4906cc27852 : SUCCESS in 3m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/41bee253426c489da6bc176c6aa0b7ca : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/161e9bed0d2c4c4cbc93507f2fcb8d77 : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/192db222fc184fa5a9f54091af181afe : SUCCESS in 21m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c0c47957118e4a3082d548098ca5c1b9 : SUCCESS in 6m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6b5368a990dc4f738524a188077ce243 : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b54b46aabb464e7ca80a3c688e2007d5 : SUCCESS in 3m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f508ccc71624c87b4c58ba0ec5fe6e9 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9d2c39461384a30b3ec2540b0100960 : SUCCESS in 26m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b1df32372cb44581a808d6d667762402 : SUCCESS in 4m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4c791e1c3e6440798d832807607dfa3f : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a66bcaa50b8418da90d154820752f1e : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3fac535bfdc4c0cb9f2b2b86b418985 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d67b408a4fe547b5a76dcf55e26acf33 : SUCCESS in 4m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58d77ee20b5b44e49c1260c84d33952e : SUCCESS in 33m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03f583da5cdc4b579b1fb861748ba221 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6f9e9df466b66665cc2b5770c45b1fc5791f0e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-28 08:17:39.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"596c9198d790c0dbc69c3c7eb847951e7eaff1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 09:13:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/017bfdd8d0534eb89e3a3dedc9d39802\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f949c1be78c0412d8833bbbdccf98661 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e20f5828d084ee2b540feb9706d2f5c : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e8cad994db848c390e41f7d02435eef : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/656b6c4f02cf481fadbb41f8b541dbd6 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0510b3d593b4472d8b0827db4ae40271 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/502f08354d1841358dd9686ba0e61240 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0de0578faa1e45e98add3dfc6f87eff9 : SUCCESS in 3m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/947d011dd13449778afc49d310e3beba : SUCCESS in 5m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0a8a0877872d4cc29f45365b9982f197 : FAILURE in 4m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/70c880d0625f4723bbb76b40c3f44b76 : FAILURE in 3m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e8d33235033c4fdc8def70f7f1d2461c : FAILURE in 3m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/215e7d7d6c89452795cea1a3599c1aca : SUCCESS in 4m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0ad7d4916ec54412b3375093e13300be : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/03cbcb1739fd4ba1a1ecfbfa4ce18c6d : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cb308a38a51c4684ba41384f4aaa50a8 : SUCCESS in 10m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7aa34ab025c34e1d953650a6c5be9bd7 : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/95a6b3172bc847f3a5061d30b7b29a20 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fce35307117e44728bb489964a5ebc41 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/516f96ec24604826886df2aae13c7625 : SUCCESS in 18m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cafd6843477c46a38b3be0a1f4d9c15c : SUCCESS in 7m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a0eb6b11b83a4998a866f5347071e62c : FAILURE in 4m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c5ac5ea537854bca88a1497dfbd5b0b4 : SUCCESS in 10m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/25104a982fbd49c0b11613e2f6950235 : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1ff496a087964a1eac7952c87aab106a : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c7545c76cdc3435b92e9ef64c0460166 : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/73f262de001245b68b5bd072ccca0dd2 : SUCCESS in 3m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/43de1254b9bb46fbb895e8ec1ba04ab7 : SUCCESS in 10m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/177c89d0d9e446afa0abbd6acff0759a : SUCCESS in 20m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0c2a525c86964b558ca9cb039451dee1 : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7c44e73d0e3d40cdb15d4a3a8dd112fe : SUCCESS in 16m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4d48fa77cb4e493a8c879648446ac034 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/faf604a5e11441bea82b778b22ed06b8 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca954b76b67449aca3392eee06c2b886 : SUCCESS in 23m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3ece3f82da4445b09642715f268df424 : SUCCESS in 7m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a8165e1ce04c422b83170d0b8ff09170 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37a21fd9515d47eaa406e2ed9c7d0524 : SUCCESS in 4m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/abbc40233e0245d7a07ff88daeb83635 : SUCCESS in 7m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/448bf9ad4f2d48e7afec3c4e9c4ea96a : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd1e67d4ba0f4a88a012a8e4214e37c8 : SUCCESS in 35m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aca6a329453d4db696a9c2a1d23c429e : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":5},{"id":"bb30e5dca6455d1929fc464b0815163429de2b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 17:54:59.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5b741a5fe674cbaae4139d9201298df648601319","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 17:55:36.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"584de1de6a764146c7228a61d7915f40d6329695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 19:44:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5805281afadf409f9b080a7933fce130\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe7417d029648e989ea2ad5a97d7a12 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c777da4226f42efbc4061dff112342e : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc3b8ba4eea74c3998887b97c4b5c358 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3af48d3f835c40b38854ba1d190fc0c8 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15ecd0e8484242d7a926489d12bc83b7 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa192b7a7ee04e52aed44f6071a926d6 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8457c0cb4e7449b290dda96a688c97db : SUCCESS in 2m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fc4dd8deb79c4bb28c92edc9f3fdfc05 : SUCCESS in 5m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a12aaa5f4cc74f28b4093c360c4f96dc : SUCCESS in 5m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/72d93f95906845959ab2b891f182d9b2 : SUCCESS in 4m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/31d8de15945641cc99fa969e465f7260 : SUCCESS in 4m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c5cbcde430d244059292940133d8a217 : SUCCESS in 4m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5cd81077ee8544eda03e4f65c2b4a2ac : SUCCESS in 5m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6bdc8106c48148e0ba237bf303c580fa : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/93da76140af1490d95df47767c39a1dd : SUCCESS in 8m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bc18f670d8224236aba1f047a562fc3a : SUCCESS in 4m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/277a23767f95401b97bd942f5007a0c8 : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/84e73654c4dc478e81a4609ad09cd32b : SUCCESS in 7m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/36b971b404034f57b8683dffbd6167cb : SUCCESS in 15m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bee7a6884d314ac4a225f49adbe3ba60 : SUCCESS in 11m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/53910eb124bb4dcc8bd73f50f173ea91 : SUCCESS in 7m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/76004458a4b243c5a1ad4c240d4eca57 : SUCCESS in 10m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/232f1641ca0945fdaf567c26deb46a9e : SUCCESS in 3m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1c07385eb3914998a9d7133420d3cbc7 : SUCCESS in 10m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/54fc120e8afe49c0ba7e4d3c8f4cccc0 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eeb26f4cc9a54af18be93a5dc54f3e9a : SUCCESS in 6m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/71c129a90a834fbe9bfa44f2fef9e286 : SUCCESS in 9m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1024dbdfc8d64cb6a34dfb6768287068 : SUCCESS in 27m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/66490a0811624f5e9d20a8453194f57a : SUCCESS in 9m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e1c6a882454b4e59b6a0a88bd96a5f6d : SUCCESS in 10m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/42376ba8ac954295a3eefd356b26f507 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b45ba249f5cb4df1974d0815567d9c4b : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/354af23661184542bb78de519fc46235 : SUCCESS in 27m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/91811f9da3cb41c6bb064a00f63452f6 : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c09422d71e7247b5bc71fe30de229327 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45fc38f1bebb44fbacda8c3d445e2c55 : SUCCESS in 6m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ee09721df2f64f16bb4e5f1f608ab61a : SUCCESS in 20m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/01a9f965ec83421f962c895f674b6f73 : SUCCESS in 6m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f86c69f14fd6489a998d989bb616f17c : SUCCESS in 1h 01m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fb539c3f69144cb934366e1c7d7425f : SUCCESS in 1h 44m 20s","accounts_in_message":[],"_revision_number":6},{"id":"7b20db2fe9f8cb2105ad6e0a90b0cd4d27926cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 11:39:21.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fa800289c6838f8b20a1dda54ec7f1beb4cba2c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 11:39:41.000000000","message":"Topic xy removed","accounts_in_message":[],"_revision_number":7},{"id":"ec2b9e10c627c630fe9366ceea6609b0d6760bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 13:38:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84a94762c50446428119f3190397dc8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f4c0427cd647afa2896faca7f7be34 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2542df4b877e4cd7a5057736fd4bf3da : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936c9fddf1bc4e5f911634835755beb4 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9a4acad199f4322af378fd3cee6ba86 : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0dd9db97c92747809a17bdb17127ec13 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38a8c84e2d214b4ba6285c4a9bd2882f : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45170a6974374079b46d2506961fecc5 : SUCCESS in 2m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f12d414124d545bf8356a08c60a4e2ef : SUCCESS in 2m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4191ed24c8d8489ba1e866be85e4f709 : SUCCESS in 5m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/28d022aeec5449ebb9cf93f4e6cb1728 : SUCCESS in 6m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a4f646d97fc949eea4cb0831d1cddb8f : SUCCESS in 5m 09s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3e3d15ca1a8f45869380ec2128beaaf7 : SUCCESS in 6m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9e9dfaae1c2d4c209caf7396270c7c0d : SUCCESS in 6m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9d4b24e470f24dd2867bedab254c408d : SUCCESS in 3m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3d8ed5b96d84404686731a4d6ffa680a : SUCCESS in 11m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/063aa9ef6a2442d79c59562ea9693020 : SUCCESS in 5m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d8bde0e09e794d62b4a6bfea168db881 : SUCCESS in 6m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/768d1cd947844a5caa7d52297b0db55d : SUCCESS in 8m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8e48e893d17346e485ef906865508443 : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b59036691c54ed1a48f3e1980c8c229 : SUCCESS in 10m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b998afb6e61e43f4973adc533cf3626d : SUCCESS in 8m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fdcc991113794d1f99f3c93e00f55c46 : SUCCESS in 9m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/483e136d2c89437aa59ff086eb34bf00 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/00cfbfa68d1c4c3f9bbd2b44505ae65e : SUCCESS in 10m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7231177769a04f44be802d8564021ec8 : SUCCESS in 2m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f4c5c3d2ebf74421bd1a2a2f382d32d2 : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8651a25074e64c149c0004736f0463a6 : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7e76715d5ad54af686432b5f80aff656 : SUCCESS in 18m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6b95a8ee9e724f608092c2e4c592f71f : SUCCESS in 7m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4ef0559c88c6478dbb96530c95359881 : SUCCESS in 16m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c480889a38144041982f82a93a99337c : SUCCESS in 2m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46a971aa2d9840a69baac808bc62cdc8 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/814f9adb702e45528c3b517b77508fd2 : SUCCESS in 23m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d65d014862c14cdbb8bcbbc729451fc0 : SUCCESS in 5m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/af7e35f0b9554cdd8adff6d48ae622ae : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65f8347a4a854a9696395f0d2cac99f6 : SUCCESS in 6m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/147ff9ceaba84007963d9e8c42989ea2 : SUCCESS in 11m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6544411f06924d21a3f59b923f2640ff : SUCCESS in 7m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86cdc31c91e34494a1c25c15b5b3f4e4 : SUCCESS in 1h 03m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6c59c2d4ef84074938e8159519ae28f : SUCCESS in 1h 53m 28s","accounts_in_message":[],"_revision_number":7},{"id":"40d7d3f595ee15db9199476270a2dac00481c35a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 15:43:45.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"eb8e48ae26aaa53a1f8318d8f4cdc96de5ed2aec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:30:21.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7e3957aae69a96d430129e1773e2c35bc453cae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 05:30:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"aacee45932a5c543a1eeeb1d1254b446aa3bdc5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:14:02.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829db565abdf402fa6c3919e8239a664\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86025213233245cc971cd8ee227e2079 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60747e07a064470cbe2b497f204611a6 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d5d34f494374ad0bc8cdaaaae69cada : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70b557eff7094864859e1ae53c297df3 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75464639ab8a4df9affc812b918eeeb6 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/666469443fbb4cf785d7c29bf9c13b52 : SUCCESS in 3m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ca33d1ea9ef3427f89f7e0d16ef36955 : SUCCESS in 5m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b02c47c2946461d95fc9048a2f61461 : SUCCESS in 3m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4267fbdc81b84afebd761ef3356b3bb6 : SUCCESS in 6m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/957c47d89a2c4e49ad14bd92bb9ae555 : SUCCESS in 5m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/33009781bc5040008464c2e16ac77157 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5e2782891c4548b1b517434b60b53f7f : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/12411c0aadd7403886db1f7b7dc36077 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5f5617fe7383493b8dc260099856a1cc : SUCCESS in 6m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/42dde672325a49e882da21df10e36c7f : SUCCESS in 17m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a1766e0800f4123ab612e6356980e3a : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ef6d9aae99c144588a66a9c2a2d76c48 : SUCCESS in 5m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36e38cadd0dc48a49b06a58ecec33a8f : SUCCESS in 12m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/61ace8730c414f48a67ab7a39edcfd5b : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a4a9ab4c41624d1798b3e3521b18413f : SUCCESS in 6m 31s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4fe1ae7fa9194e0b9dffff9ce21d2650 : SUCCESS in 18m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f7344b0062784b249daf510334edd6a2 : SUCCESS in 5m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f371161f3b68435bad9e7e9f1cf8f4a9 : SUCCESS in 14m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/583ebcb5e99d474cb949c936ec78dff4 : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b673c618f463479586de76f1353c8a9c : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3935f180970e47e8bde6cfc2e2e9d4ba : SUCCESS in 24m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/becbb343153e43bbb7ebfb033bc32b54 : SUCCESS in 6m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0800129c8f3a4e5a9250c96fcbff3fda : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b79fcff60af74afaa7b87b4f3d32578e : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/15dc7510cd0b450494f835a658dd3cc7 : SUCCESS in 12m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7aafee6367434a95a45e57846534caac : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76a95147c50347a9ba76d0f4f9df3f03 : SUCCESS in 51m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57581b5fc30b4477846e0bf06b4b0696 : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":7},{"id":"d6458c4f84de3a644ca763bb88395ec751561473","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:14:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"dabc4daf67c3a602c8e2d2fdedd6769aa6fdfd34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:14:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4aba9755c144ef2b526d4abd30d7f17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8681fc3dc549f08a515a319e248dac : SUCCESS in 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"12c21da56867f1ca4746b1f351194b23ce55b802","revisions":{"7e054e87478f71acfecb00dd01338e95390c8ba6":{"kind":"REWORK","_number":1,"created":"2026-02-28 08:15:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/978295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/1"}}},"commit":{"parents":[{"commit":"271adef517e153b5899a4689151318c44085a908","subject":"Merge \"update constraint for python-barbicanclient to new release 7.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271adef517e153b5899a4689151318c44085a908"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-28 08:15:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-28 08:15:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e054e87478f71acfecb00dd01338e95390c8ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e054e87478f71acfecb00dd01338e95390c8ba6"}]},"branch":"refs/heads/master"},"855c649f750e3dc0b69a26ffdb08284e09cf5d4f":{"kind":"REWORK","_number":2,"created":"2026-03-07 08:16:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/978295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/2"}}},"commit":{"parents":[{"commit":"9de80aedebed1e83c8a35a0a94ace6efc381224f","subject":"update constraint for castellan to new release 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9de80aedebed1e83c8a35a0a94ace6efc381224f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-07 08:16:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-07 08:16:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/855c649f750e3dc0b69a26ffdb08284e09cf5d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/855c649f750e3dc0b69a26ffdb08284e09cf5d4f"}]},"branch":"refs/heads/master"},"82bc1dfdbf0ce5efbe09e54086d0f6c6cbef4955":{"kind":"REWORK","_number":3,"created":"2026-03-14 08:09:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/978295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/3"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-14 08:09:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-14 08:09:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82bc1dfdbf0ce5efbe09e54086d0f6c6cbef4955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82bc1dfdbf0ce5efbe09e54086d0f6c6cbef4955"}]},"branch":"refs/heads/master"},"34f4dd62a33632369d77dfb2d3d6786066c53a65":{"kind":"REWORK","_number":4,"created":"2026-03-21 08:19:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/978295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/4"}}},"commit":{"parents":[{"commit":"754ae37e67ba08c1695698ea7a8ebe52efa711d9","subject":"Merge \"update constraint for horizon to new release 25.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/754ae37e67ba08c1695698ea7a8ebe52efa711d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-21 08:19:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-21 08:19:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34f4dd62a33632369d77dfb2d3d6786066c53a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34f4dd62a33632369d77dfb2d3d6786066c53a65"}]},"branch":"refs/heads/master"},"aab3957b7b830da5f1cca934cb3043462ce76b9e":{"kind":"REWORK","_number":5,"created":"2026-03-28 08:17:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/978295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/5"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-28 08:17:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-28 08:17:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aab3957b7b830da5f1cca934cb3043462ce76b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aab3957b7b830da5f1cca934cb3043462ce76b9e"}]},"branch":"refs/heads/master"},"6e97c08f6d1dcdd4f5bb201722f32fc8b9039382":{"kind":"REWORK","_number":6,"created":"2026-04-01 17:54:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/978295/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/6"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-28 08:17:38.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-01 17:54:01.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e97c08f6d1dcdd4f5bb201722f32fc8b9039382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e97c08f6d1dcdd4f5bb201722f32fc8b9039382"}]},"branch":"refs/heads/master"},"12c21da56867f1ca4746b1f351194b23ce55b802":{"kind":"REWORK","_number":7,"created":"2026-04-02 11:39:21.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/95/978295/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/978295/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/978295/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/978295/7"}}},"commit":{"parents":[{"commit":"d98ab6073f2cba9c1c3b9dc21c91b32e601aeeae","subject":"Merge \"update constraint for futurist to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d98ab6073f2cba9c1c3b9dc21c91b32e601aeeae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-28 08:17:38.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-02 11:38:44.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- pyroute2: issues with neutron\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ifdce39f736a5a61586e304c8978d4d159ab0bbd2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12c21da56867f1ca4746b1f351194b23ce55b802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12c21da56867f1ca4746b1f351194b23ce55b802"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
