)]}'
{"id":"openstack%2Frequirements~978428","triplet_id":"openstack%2Frequirements~master~Ic8ac321136cb9333267f6bd3031980cf5c4d4188","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic8ac321136cb9333267f6bd3031980cf5c4d4188","subject":"update constraint for python-cinderclient to new release 9.9.0","status":"MERGED","created":"2026-03-02 10:28:00.000000000","updated":"2026-03-03 08:54:45.000000000","submitted":"2026-03-03 08:52:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978428-new-release","meta_rev_id":"f0db85a06a893c11ef582383ff67e9743fcb9743","_number":978428,"virtual_id_number":978428,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 08:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-03-02 14:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-03-03 06:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-03-03 06:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 12:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 14:25:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-03-03 06:54:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ecb4040f4b365e61e7a01e5678605c607608d3b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-02 10:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"418cedaa9e13c579e5b14f01f9a7b777a7510576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 12:32:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be4da60fc56453eac005196cb6d30c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3699c8fe5eb4be1a8966176dbbdd374 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9ffd72d726401b8c361d160ba3ee01 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d559434294044fae92a4d951cec0a0ab : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/882a248ff6e442d8ac20f7ba97a80706 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f6e94084dbd4c16af70846e43a44b24 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/289b657c556a44b0be6dd9502a6fd5de : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0578578e5b1d47ba927f08f72e80c775 : SUCCESS in 2m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6fc1f4fce1794f4fb1031f213aef3288 : SUCCESS in 6m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0a6d85d7d4a24c149519b14fbaeecb4f : SUCCESS in 6m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0f3187a032e741a78705d356f1599eb7 : SUCCESS in 6m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/20c822c23db6487787eae3d43eb7f4c8 : SUCCESS in 4m 53s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f65256611d5f422fb1feb42982b1f671 : SUCCESS in 5m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c91739c061f44ff198d70ebe386818e7 : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4501d90c2a7e4d8a953b9403e66eb255 : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7871e1ad1d73483386d7a0e3ffed1359 : SUCCESS in 8m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0f378d2d472b46d5875e984279db8c7f : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f716c360763458a9b58f82aeb8b88b3 : SUCCESS in 3m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6b8f309bb9c5493ba8358489e5d3aef1 : SUCCESS in 4m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ed920be2bf7c4929ae85c7c657e24367 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e6bfe2d12e2436281e9e0872a9cdcd4 : SUCCESS in 6m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/74ea38e89e53448ea5e733d7ff7f3781 : SUCCESS in 8m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f44e0efd7dce462f87df5c8f871ee17b : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5b2e5d0fc6b94c9d8c6e9570b3c8b83a : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e20b7a6558694872a5ee9313c002a94a : SUCCESS in 10m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8bd6cd40014546c9b9b5cede11b58210 : SUCCESS in 2m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/639fd21573224bb88a40a9a21fa0780f : SUCCESS in 2m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3b2ae84076bf4c2e9f31478d97d8476e : SUCCESS in 10m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/adfea3e671414464896ab2e58581a658 : SUCCESS in 18m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2710606300494a5891104ca605a96f34 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d8af207434ef4d57998b773f7a8e4488 : SUCCESS in 11m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6eecadc647de4ee283cca86c34a3f9f6 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a123ceb0ae924863aca6a16df2dd8d89 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb56b5626d444a91afe6b479eb5393a5 : SUCCESS in 27m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/44874f5f78e24269aa9f6375aad1f5d5 : SUCCESS in 7m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7890bf388881480ca2b6b5d0c733f1e6 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b9670eabcf142c38b1f08348d606d27 : SUCCESS in 6m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/de0bce86cdfb43f98fa4b8a41259c4d3 : SUCCESS in 19m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0c8c51f906844043a16d0c43651fb4ae : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc0aa95bdb5d4a0ba02196f7e3b08e15 : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9443c8a5c3ca4734a0b0ab738d9de3f1 : SUCCESS in 1h 46m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e868dbb71a0951e6f76e15d864dab8bdf48ae131","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-02 14:25:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"505d6b5ed2c67f1659e7c40029c43646659d0378","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-03 06:54:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aecd6e811e8497d3691ab066502ef7c19a78b443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:55:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2bf4624b40286fb58487b109f74174f4b36d53a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 08:52:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15ec788e8617411784441235c9249413\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333ecbc53d67436fba9fe4921f08707e : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebc83bce2104f9190aca4cca95c1e7b : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bd8e78026584621839434af53a67cb1 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cbc9652ff3249a491f83bd27adc4de2 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fefdb09ceeb34a7e916e5331f8aec403 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e38c5c7ec27648a59b21505054a844a4 : SUCCESS in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2a1332607a0b4232afb7ca332a345c08 : SUCCESS in 7m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ec31c8de85334800817ab1a48448ab46 : SUCCESS in 6m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/de3982e372b442c6bcc68776313168ad : SUCCESS in 5m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/31566771c0094b28a14f0f4125549b81 : SUCCESS in 6m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/00526eb4ebf7458eb5e9e8295ed36908 : SUCCESS in 12m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1a5455b333d842eaa54aee2461d1270f : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/99949e5b256449abba59e3cfd27d7b1f : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ede047dc9a764f009379e3e0fe12872c : SUCCESS in 7m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b211a10c01ca494a8df9a0747fdd1cda : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61e1487f83e1410fb8fae45f9d1d0afa : SUCCESS in 7m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e0f99709a8bd46f98720ef65821c9ab9 : SUCCESS in 7m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/234ca85815f048c8a3a0f6b62821b0e3 : SUCCESS in 9m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8197084001954d3f9c45a906132878f2 : SUCCESS in 1m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/db8aa96e2e6b433b97a5658186e40331 : SUCCESS in 6m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3efca8ae29f34b979b0ad7e7c1e86e91 : SUCCESS in 28m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a26e7aa6b45e489faa5a77e933582270 : SUCCESS in 6m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4f6a02affc1c46e7a125d6ba9a0d2871 : SUCCESS in 15m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f461268ad55240d4a84a4cf212acd196 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09cc7f480deb4790a0bae44f79938c93 : SUCCESS in 2m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26298cbc25ab4edc8c1569bf83c4876c : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b2e36d3fc1d1496c9829c8fdb5ae618a : SUCCESS in 6m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a816450bb366424ebb4d3629fcfc8a81 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b67e8d93dae4c5ab4aefb2e5130664d : SUCCESS in 5m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0ee9a2521d76420ea81d128db446075a : SUCCESS in 7m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dc753d1c2102486b887c7cc9cdb5f9b3 : SUCCESS in 3m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d122d3e6baf94f48a4d15262105a61a9 : SUCCESS in 1h 00m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12cbc388e36741309bee306a329beaaf : SUCCESS in 1h 53m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bdac7281f7c4ad6937bc114131144b9d55ff6220","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 08:52:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f0db85a06a893c11ef582383ff67e9743fcb9743","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 08:54:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b339ce6839dc42e69ded6c02a5c4f58a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca11b28043e4d94879bf67f23fa6fc7 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c702fcd2637448c8a69b9a24857cd519ddde4952","revisions":{"c702fcd2637448c8a69b9a24857cd519ddde4952":{"kind":"REWORK","_number":1,"created":"2026-03-02 10:28:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/978428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/978428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/978428/1"}}},"commit":{"parents":[{"commit":"6743428d09acbbbb92ba5b7d1bdc003ea4bbdba4","subject":"Merge \"update constraint for python-vitrageclient to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6743428d09acbbbb92ba5b7d1bdc003ea4bbdba4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-02 10:17:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-02 10:17:58.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 9.9.0","message":"update constraint for python-cinderclient to new release 9.9.0\n\nmeta: version: 9.9.0\nmeta: team: cinder\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: Ice888623f8eac3bc61e01252f6eaaa5a4fe7c0cc\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic8ac321136cb9333267f6bd3031980cf5c4d4188\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c702fcd2637448c8a69b9a24857cd519ddde4952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c702fcd2637448c8a69b9a24857cd519ddde4952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
