)]}'
{"id":"openstack%2Frequirements~978679","triplet_id":"openstack%2Frequirements~master~I651dc04b2cc2ba3b603acce104015ffba464e0e5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I651dc04b2cc2ba3b603acce104015ffba464e0e5","subject":"update constraint for networking-generic-switch to new release 9.0.0","status":"MERGED","created":"2026-03-03 09:13:10.000000000","updated":"2026-03-04 10:46:59.000000000","submitted":"2026-03-04 10:45:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978679-new-release","meta_rev_id":"0e3fbaa33a6aadfb8db104089a6d015a0441901d","_number":978679,"virtual_id_number":978679,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-04 10:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-03-04 08:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-03-04 09:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-03-04 08:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 11:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 08:42:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-04 09:11:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"5d0dea499fa4885112f26a9116009a87f81f959f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-03 09:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c17d580b05a603618e0d5debd086d4488460aaef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:07:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8fd56871b6420f8cc3d9d4e5bca95e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f913fca94d46a986cf6557b5fbe38e : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509984c94f7146f682b17dc5fecdf994 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1f8f3eaa57d49b09f0e807b9509d3e7 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b7637bf529a4ded8b3b389bf381fa0a : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb72e7ed76c94926b20eaa2ff1c8192c : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/523ba6549b2b4f02beb44d5fba2e37f8 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5f6246aaa1249ec8f55524e38cd5a8b : SUCCESS in 2m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e39a65a0b7e443aa3d9fa75b5f03f5f : SUCCESS in 4m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/48302212f61149deabb8a6459a88e3be : SUCCESS in 9m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2ddb293020b94d188b92cce74a51e51e : SUCCESS in 4m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5d565a13979e4fd4b7fa26f14eb95ae5 : SUCCESS in 7m 58s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/440494f4ce3d41d1837cae670147c725 : SUCCESS in 5m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ade637635b5d46009baf2817a7aa2946 : SUCCESS in 6m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8ee64001da3245efb6dd4b0eccb94f21 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e8467d5c8a254aa98763105505a55b0b : SUCCESS in 7m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e74fa578b2b3480cb529189436128385 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a21af2c4056e41aa9f645afc1d5566ce : SUCCESS in 8m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/be1032a884a948788b92e04e007620ba : SUCCESS in 6m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/56b5d49d61674293bf0896416ad85d9c : SUCCESS in 18m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d197fdf62a0f416ba6d4ba5e2f20f636 : SUCCESS in 5m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/379224b924a44d04a8fd2f9d3bf9837b : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2662ea433999487582a9a3912dc858ea : SUCCESS in 11m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2f4a10cd0531423b9f18e00870967fd6 : SUCCESS in 1m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/541a8e54dd02406dac67d4539c474c0c : SUCCESS in 6m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2eddb9486f63492aa403a01f8c71da54 : SUCCESS in 5m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bd9676fb6ca84836a6eeab9b8df6c6a2 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cc2b851ed2d942b5b3e0205fb998e73b : SUCCESS in 10m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/34755edc9adc455bbb1bdca13e0b0090 : SUCCESS in 24m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dd408a38476443aa9b5b309b3cf5337c : SUCCESS in 8m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2d6d66a8a48b4afe80fe5feaf2ad83f1 : SUCCESS in 16m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/87fb01a6720f42d19125864947eb199e : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d6cf7aebf31472d9c7af8b4b34798ec : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8729ef6e91f7415e824e6b4f05f0870a : SUCCESS in 28m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/47e14545eef849c1bf5e8e7eca2075f0 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4310aae4769841d5b8b66683abfd4acc : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d7d6e2540854b49b503b66ae0233de0 : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/455268091b724ad094eb3dc4c4ec9782 : SUCCESS in 7m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/975f2d8761bf47b5bb613d59808add0d : SUCCESS in 6m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e39cce13499a401ebc34d0679408ed93 : SUCCESS in 33m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5daf2c5ca9a846a8ace2c1ea2a649176 : SUCCESS in 1h 46m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9ab08d0d56e1bf64482e3268cd0b526f72785f2f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-04 08:42:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6edbb0551b44bbe82ff5460deb4a17c4a37e55f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 08:43:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b441fa8f5f9dd51b1e1ece9e864f361122c7b95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-04 09:11:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f450ecc332a45d14e70e534487f5628f86e6e70b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:45:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d5f2f5a2f645ae8cc886ecfa87f21f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a64e338a6346f4be4f808f96602e88 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303a4b2c39e949aba8bfef96e5eaae18 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86c2f384c4484b29af09e63971555f7b : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/680aa8a4c7c346bb84b9bf99fceebb23 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/203364807a00402f9ed8ea7b18b19214 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ebfc734378741e5ac366ca4478b3935 : SUCCESS in 5m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cee0c3c9b2be4be6a9a2ced612f6b6be : SUCCESS in 4m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/11a9848f9fec44cc9cb5e74860f376e3 : SUCCESS in 4m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/832c9f67cc9e490cac30d858129cbb52 : SUCCESS in 6m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/359ff9a3f2434fa2adeff6acf8155dce : SUCCESS in 5m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c7c3db143a03421ba4ec2818b28e2a37 : SUCCESS in 12m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/daf38f3e4e0c40038e68c3d12852245e : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dc4faaf9713b46c1bc942bcd57faf511 : SUCCESS in 7m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aaa249566762485bb8a080ee91cb1ef6 : SUCCESS in 7m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5305b33f56f4448abe7c5b699cde040c : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ae587f16a104aa6841c71b26d7e34c4 : SUCCESS in 6m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a842d67d17f8432990869a0ccfffdde8 : SUCCESS in 9m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/49327eccfa3d4ffdbc69627cc0e834a4 : SUCCESS in 10m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aeaec8219d82435ca5fb510f9f13b621 : SUCCESS in 2m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8fdd9930c439407fb296da6370b53b2f : SUCCESS in 11m 01s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1909e896d80d4b29accdbe4ebe925ace : SUCCESS in 18m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/77d859e086984e32ac720db5d0c8f9a7 : SUCCESS in 5m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/28434607d1d7436681e0c67a0d3706e0 : SUCCESS in 18m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7437eac90d26492e88a38848e0872058 : SUCCESS in 3m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2127dfa2925e4550854cf35e160b3432 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc38de5d84df4db2a26c4cff0d2ee5b4 : SUCCESS in 22m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2455002f0fc5495e8fabfbad202db09b : SUCCESS in 9m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/27c02b46685445aabc41e3ac50f8855c : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/038f20ffbf964acaa7784a746322c525 : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/953f9f70fc584f8cb6d69ce7d9a4db0f : SUCCESS in 13m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0fa547ba23af4d1c8e942fdf7da52272 : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0258fc3cb804ad28003778af1e1b3f5 : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1564162232df45e1b0771624a9d20586 : SUCCESS in 1h 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"964a4b62329ac9570c801139da24c5749174ade2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:45:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e3fbaa33a6aadfb8db104089a6d015a0441901d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:46:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0cfdfb2cb8b4f17b1b2187a8808728d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ffabe9beac4162949fb02a0ee02907 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"909226bc79180af4d676156addfc7a12bea21279","revisions":{"909226bc79180af4d676156addfc7a12bea21279":{"kind":"REWORK","_number":1,"created":"2026-03-03 09:13:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/978679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/978679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/978679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/978679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/978679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/978679/1"}}},"commit":{"parents":[{"commit":"c702fcd2637448c8a69b9a24857cd519ddde4952","subject":"update constraint for python-cinderclient to new release 9.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c702fcd2637448c8a69b9a24857cd519ddde4952"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-03 09:03:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-03 09:03:08.000000000","tz":0},"subject":"update constraint for networking-generic-switch to new release 9.0.0","message":"update constraint for networking-generic-switch to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I437a4d7d3fe454409b4de59f23405a38f4aecb1a\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I651dc04b2cc2ba3b603acce104015ffba464e0e5\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/909226bc79180af4d676156addfc7a12bea21279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/909226bc79180af4d676156addfc7a12bea21279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
