)]}'
{"id":"openstack%2Frequirements~978740","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I39d2c867476106c7127a2365d31713c8d07abb38","project":"openstack/requirements","branch":"stable/2025.1","attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-03-03 12:56:58.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I39d2c867476106c7127a2365d31713c8d07abb38","subject":"update constraints for various XStatic-* releases","status":"NEW","created":"2026-03-03 11:47:53.000000000","updated":"2026-03-04 08:51:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fcbf7e06ff61f5ecdb24d01b52fd91f58f5cb251","_number":978740,"virtual_id_number":978740,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-03 12:56:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 12:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 07:54:03.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"CC"},{"updated":"2026-03-04 08:51:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"377b6ca23eecfd1b16f33d2361a3be59c536154a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-03 11:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9460b34a5ca792776627d7289f9dab035e3a9944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:56:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d826aed00c04c26aa0074b6c5f7ed9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cc4315e546402f9ecc3639928f4622 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e9205be7c79427f894d754db21c86d2 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9053e2f2009d4c7db0a545ed917d91c3 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bcde8d733f9a45d784c18db50b41c314 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc3b51ecce47415988d1d41baab69f25 : SUCCESS in 2m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f75650d2fdb4007b910348567a7d486 : FAILURE in 18m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d7f672e9823d4d89b81b89abd97f984e : SUCCESS in 15m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c4de3e7474e3474da3cfe319c321b9f9 : SUCCESS in 17m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/196a8c74ddc74012866a19f2422da19d : SUCCESS in 18m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/00f5ac52242c426aa3f91aa41f3f1f1b : SUCCESS in 5m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a7171a97e26b45968dbe49408718a7c6 : SUCCESS in 3m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9856314452864f7186b5f963a05aa361 : SUCCESS in 5m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fd8e75a9fd7e47198916c05f875fe966 : FAILURE in 5m 57s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a023fad459464cc096ec89225119954a : SUCCESS in 7m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d1ccf8bf81044d8bb47af254bc49c543 : SUCCESS in 3m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/085b9f1df5034cf38c6d528429ac1dee : SUCCESS in 7m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9ddd5ea464e845ccb165bef61c8acd48 : SUCCESS in 6m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/011ed3328d54425f8da5351377f7263a : SUCCESS in 9m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d2afadd1db3f44adb9773c48c5f9ef35 : SUCCESS in 9m 48s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/96495934dc8249d09417a43beedf81d9 : FAILURE in 4m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e434f9fa3d274f51a33f0fb6fb0ed394 : SUCCESS in 2m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/574f2326e3f1401c92f9232a79e32919 : SUCCESS in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7f28b80f61ce4f25b1d1392824b3ac26 : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6442df4954b64a9d97eee282f76da6dc : FAILURE in 3m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f77a6ba228a74e43878d2e36c7f0989e : FAILURE in 5m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b4ee99506f17464b88292afc190b1f51 : SUCCESS in 23m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/59da59ac720b4ece8584fc2b002a7488 : SUCCESS in 9m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/18c4fde9bdcf412cb5610041a8c888cb : SUCCESS in 12m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7641cb805a941c797dda939a316546f : SUCCESS in 3m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/213cbf63535942f689cca4171a8a79b2 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a647ae142ac64427898817d2e7bb5d44 : SUCCESS in 23m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f3020953587f4ba0a255707cfb9b971d : SUCCESS in 4m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0824bf31e2be46df9806ace0bc259f95 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e35c5f464a0e4479ad91cec8672de989 : FAILURE in 3m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/505f473af8f843f98cd90a89e091383b : SUCCESS in 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e77d967619a436d841f889c4bc64d41 : SUCCESS in 1h 03m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b37b9f58398b480dbe0b8e6c41e54be1 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fcbf7e06ff61f5ecdb24d01b52fd91f58f5cb251","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-04 08:51:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2c8a7bf5881beb92d2fa3765ed9ffbff3eb7355","revisions":{"f2c8a7bf5881beb92d2fa3765ed9ffbff3eb7355":{"kind":"REWORK","_number":1,"created":"2026-03-03 11:47:53.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/40/978740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/978740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/978740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/978740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/978740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/978740/1"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-03-03 11:41:53.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-03-03 11:47:47.000000000","tz":120},"subject":"update constraints for various XStatic-* releases","message":"update constraints for various XStatic-* releases\n\nThis is a squash of multiple bot generated patches, they are all needed\ntogether in order to make the cross-horizon jobs pass again after the\nlatest setuptools release. These all correspond to two release patches\nwhose metadata is kept below minus the actual version information.\n\nIn addition the externally updated XStatic-term.js lib is bumped to a\nfixed version.\n\nFurther fixes in order to get the CI passing:\n\nDrop vitrage related xstatic libs and fix CI\n\nThe vitrage project is inactive, so we cannot properly update these to\nfix the breakage caused by the latest setuptools version.\n\nAlso make the cinder and ironic jobs non-voting as they are currently\npermanently failing.\n\nCo-Authored: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I39d2c867476106c7127a2365d31713c8d07abb38\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2c8a7bf5881beb92d2fa3765ed9ffbff3eb7355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2c8a7bf5881beb92d2fa3765ed9ffbff3eb7355"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
