)]}'
{"id":"openstack%2Frequirements~978871","triplet_id":"openstack%2Frequirements~master~I87fc782459631fbc4a2ef54787c4c87b91159b1a","project":"openstack/requirements","branch":"master","topic":"fixtures","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-04 18:32:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87fc782459631fbc4a2ef54787c4c87b91159b1a","subject":"Bump fixtures","status":"MERGED","created":"2026-03-04 12:29:55.000000000","updated":"2026-03-04 18:34:06.000000000","submitted":"2026-03-04 18:32:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978871-fixtures","meta_rev_id":"78747d192d2d746c15533b5ab81b8ec4c80a1ece","_number":978871,"virtual_id_number":978871,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-04 18:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2026-03-04 15:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2026-03-04 15:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-03-04 15:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-03-04 15:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-04 14:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 15:16:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2026-03-04 15:29:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-03-04 15:37:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c1b0f7c089a9de240cbdca6e0c78c35a4ffab4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-04 12:29:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82d4efbd4e88fa5567cc395a93a409df89be48d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 14:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a13948d6f044d5b9c527140a2d55202\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46df0b2765b842bd881b50249a2db87c : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2ed7a71f55434c917243f1e76e288f : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43bca7162de8495f87c53b61b4ad290e : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55fe333415aa487983bde22c6bf49a71 : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0f052889687742d98d2b1e749a15d1f1 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c45e05da7d8845eb85af1472b2859f66 : SUCCESS in 1m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/368401b2324342fcbe1d227eb14d4b29 : SUCCESS in 1m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa94e37d3d89475dad36d0f2338c7fa6 : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/faceb9db77be46cbafd49c2bb53f3a82 : SUCCESS in 7m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d77bcb02cb124f21bfc1bb324414596b : SUCCESS in 3m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/beb7487ecd974f43a6a3fe109d272166 : SUCCESS in 8m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/170ba9d2f99a45cdbcdb4ad9c789fc39 : SUCCESS in 5m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/19e9d04f6e56494e9d39ed99888ff24f : SUCCESS in 3m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ea0c2cff6fdc489dbec3f53fcc732c8b : SUCCESS in 4m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6eb71931996b4eedbbf89050f6bbf501 : SUCCESS in 7m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7478dd8c7f28448cae12af07eb7f5a12 : SUCCESS in 3m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b6ffb34a0b2e4f33bb0bcb6254074ae1 : SUCCESS in 4m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a9538e9d8051475697f7ef74670e5f58 : SUCCESS in 3m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6104cd3007514aab8e9dbda56132233f : SUCCESS in 14m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82c8631c8cb84b42b779e8c109fa14d9 : SUCCESS in 6m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ee560355b2844048a35744dbf0291dac : SUCCESS in 6m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/71616c282be54bbe92c13755d1a96460 : SUCCESS in 13m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c317152e27ff417d9b9388a0b7b1022a : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4ec3d81f11874f2ea911dc7a296adc6d : SUCCESS in 8m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bf7295cfc6de4259b6c4312382255d0c : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/272b9b12530d488eb50ec474b56fc7ee : SUCCESS in 7m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0cf55bb9713d4aa389c3ccf39a54de10 : SUCCESS in 9m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/621a126259ee40e4beb1a90c946ffa71 : SUCCESS in 29m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/61a47efac17840818692220feb10a1de : SUCCESS in 6m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/22592b741c4f46cab97bd72853051541 : SUCCESS in 12m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f05e30a20e14067a2d463565b69887e : SUCCESS in 2m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/45635c19b8eb4351b51fd864315f57be : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f49ea22cb98545f59e3610ba8e50e94f : SUCCESS in 27m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f294f85ea9af4dd79eb2604d02337149 : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c4262a69cc65466db87177e1943b37e1 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7424774da574e74b48db477f8deb725 : SUCCESS in 3m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/adfdc5f8f6b144d4a957fb521b798836 : SUCCESS in 19m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8b9842f6b1844669932004a137dc6def : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/387f25c7a7b4446db11dd32ffd51c598 : SUCCESS in 1h 04m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a58edee41abd49868776071ba2bd03c8 : SUCCESS in 1h 42m 51s","accounts_in_message":[],"_revision_number":1},{"id":"09b5c48718cb2be481e591d818f14bf49f45a118","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2026-03-04 15:16:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b827d8bfe52c8943d3b4f44e215c10e4dacb03ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-04 15:29:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10a03192997fd36ad6a32e6d4c8ecf89358c0503","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-04 15:37:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3529a33bf55f95a25a24f8f4d6888c9b7781d76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 15:38:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9cb2e2419bc6106f3dba01b50a95a11ccf0e349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 18:32:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b594bf15d444b8fb09cb59eea987104\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1433f9867448849682b4c13f6ae3dc : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a630d84b4a4b52a9dfb6876fe15cc7 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0b9e4a0672f40eb9b0aeb36b55d76be : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41625230031944299f3c634f98020040 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a27271b6e7d4507bae1a6a9452d59c7 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10dbcd47e57e4dbb9e019aaf6546a4b7 : SUCCESS in 3m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c7b4d7696f3d4df692486cc8dea660a8 : SUCCESS in 6m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/16f9986561154014b301aa916d1eade4 : SUCCESS in 8m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/43633b05995842fdbece920a948536bf : SUCCESS in 6m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cc59c1d8c0da4d3ba0a6d58b89d19d5c : SUCCESS in 5m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/30767bb5c049424a82066ccd13d1caba : SUCCESS in 13m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f3a97a1aef6b412ebf3673f3294c000f : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e08c5a642b3845718a8c76c28f64ea10 : SUCCESS in 5m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9bd9a739182943e5bf2860570ba5465d : SUCCESS in 5m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2d20cac67eed4e3f92c0bc3425f50081 : SUCCESS in 15m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6628991569404fd39dd4090d7d4af43e : SUCCESS in 5m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/48af57da82bb4d369f71363d2221a0f8 : SUCCESS in 9m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/652c4a5e251a47c68a60de7feb7ea679 : SUCCESS in 10m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/46a858948bad4568a1776337c1f2fb9a : SUCCESS in 1m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/27582939ee914feb9fcdf79ac498c21f : SUCCESS in 11m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/16611f8ca66449c0804e2a64b4fb7a0f : SUCCESS in 28m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7a0bd3e1ac74c00a9776abed837ea82 : SUCCESS in 8m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/dedbd75826914e06a5e686a0ec4d89a3 : SUCCESS in 14m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6c8c17e369134d0c807576b987aa0879 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77caec2b75284c6b8a9727e67c5b2299 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c28226d804e94b91ba79c9d00028dc63 : SUCCESS in 29m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f041a2be558242b0b841d673de72b2be : SUCCESS in 7m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8d64832bbfc34c94b9f4328802803f31 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c247eaebc5041ad8d8c79c5e55b4d8f : SUCCESS in 6m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/313faeaa29cd4cddbecf21e4b335d891 : SUCCESS in 13m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/28c5e1bc4d824c92ada5d6425b9ec000 : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0410632494046adb2eb12ac62c5c60b : SUCCESS in 33m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a5db11af41b4bfd8dfdf21bd2c72c36 : SUCCESS in 1h 55m 14s","accounts_in_message":[],"_revision_number":1},{"id":"11ba43fac54eecbb1d0d5af4e3f8ef5a0b7d3aed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 18:32:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78747d192d2d746c15533b5ab81b8ec4c80a1ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 18:34:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ccde0e7f794c2f8ea27853e5b2229b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7618bad8fc487babc738e8f54ca57e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acaba862e72e92f54780e304cac9ba400e9e6ffe","revisions":{"acaba862e72e92f54780e304cac9ba400e9e6ffe":{"kind":"REWORK","_number":1,"created":"2026-03-04 12:29:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/978871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/978871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/978871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/978871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/978871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/978871/1"}}},"commit":{"parents":[{"commit":"909226bc79180af4d676156addfc7a12bea21279","subject":"update constraint for networking-generic-switch to new release 9.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/909226bc79180af4d676156addfc7a12bea21279"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-04 12:28:58.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-04 12:28:58.000000000","tz":0},"subject":"Bump fixtures","message":"Bump fixtures\n\n4.3.0 had some typing issues that were causing CI failures when using a\nfixture as a context manager (e.g. `with fixtures.Foo():`). Bump the\nversion to correct this.\n\nChange-Id: I87fc782459631fbc4a2ef54787c4c87b91159b1a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acaba862e72e92f54780e304cac9ba400e9e6ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acaba862e72e92f54780e304cac9ba400e9e6ffe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
