)]}'
{"id":"openstack%2Frequirements~979862","triplet_id":"openstack%2Frequirements~master~Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae","project":"openstack/requirements","branch":"master","topic":"django52","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2026-05-20 09:46:31.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-05-08 02:45:07.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-05-20 07:49:45.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-20 09:46:31.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-05-20 09:46:31.000000000","reason":"Change was submitted"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-05-20 09:46:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae","subject":"Update Django to 5.2.12","status":"MERGED","created":"2026-03-10 15:00:38.000000000","updated":"2026-05-20 09:47:35.000000000","submitted":"2026-05-20 09:46:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"979862-django52","meta_rev_id":"432f899a42129f6ace378dca058a21288f6fe531","_number":979862,"virtual_id_number":979862,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 09:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-20 07:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2026-05-19 16:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-20 07:49:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-10 15:10:29.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2026-03-10 15:10:29.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-03-10 16:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 16:33:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-03-10 16:33:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-10 16:33:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2026-05-19 16:56:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f0edf0844a4006c935769a0d7352862fd7254615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-10 15:00:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9255ef9624e08983ae46211864fd16aa75751f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:22:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ced6f5fa9f0f4b5581c30b49715f7bdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553d9249d6394b268b7855ed90b3ac90 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3b259247334183a77b34e0ee13a215 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03cddfbc895c4e5eaf4cc6fb943f3dee : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39abbffc3f744dc3aa03af9458e13938 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d04bde49d5284a96afb5d4cbd1b89ba7 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ab0c8e3464746298b0705a0169d982a : SUCCESS in 2m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88cbeae4a9024fcabd22150ee1e29af6 : SUCCESS in 2m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f5c2ec47e56346349efb8c9af4415ee5 : FAILURE in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c09d4207d644696ab9685f5c98e9471 : FAILURE in 6m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/182934befa5b4194a94cca797b1afba5 : FAILURE in 3m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5dc8322f8ccc4827b765e0f06c81d9d2 : FAILURE in 7m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/62916505ffe24791a26054119439a2a5 : SUCCESS in 5m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c20969d766a445b693c3d4bbc99bb306 : SUCCESS in 3m 36s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2bc6fa96603b40be9304b2000b292ca3 : SUCCESS in 4m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d913e96c1ec642e7a186a626ea04f7b9 : SUCCESS in 14m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ee28fc496ef0465096ff333b5e692d4f : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d52ba5386de486da5f6f382f88a5cbc : SUCCESS in 4m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3f87b7a1212040ba84bca0ef2d6392ff : SUCCESS in 8m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0fb90d934a7b45be81582a735a9b6272 : FAILURE in 3m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab5e70ff8209490192ceabbe99f7df19 : FAILURE in 3m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/624ba6ff802943bb87beadf226e00e23 : SUCCESS in 5m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ed9808b7e90847b1af3ec657950a8e47 : SUCCESS in 16m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8f7fcb0da16e4ffa97919f92f866f248 : SUCCESS in 3m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0b1dee95c61149eea46e451d9550acb8 : SUCCESS in 9m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9695fa5fdd0744c7bf6fbfb0cbb900b9 : SUCCESS in 4m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7a201f5ed60b49888c93b6e2a3824029 : SUCCESS in 5m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ef43bfa703bf4bb5b44959cd396d8843 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1e7f9419288f4095b4d947f6377e2987 : SUCCESS in 31m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/48c8d44bc4b74e0987f556bb10427b7d : SUCCESS in 4m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0347dd2d1da042dea6e1c4a7497e0bf3 : SUCCESS in 12m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/32bd899b042e4dd893c2cf22e6ff63e0 : SUCCESS in 3m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4613cfc0a40478abf00600946db95d9 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f21f1602a5d4eafbf51392e795c0344 : SUCCESS in 23m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7adc155355744382b774c52ced4c97b6 : FAILURE in 3m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5e9473f90bae457b99978bc1643345a6 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90bea42be95945b5b5928d78683eca25 : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/aa646e09c5654712bb358ff01f9df380 : SUCCESS in 14m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7ded828fbd34478b96b26543e8c53020 : SUCCESS in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b301cbbaa30e41a187af73f6a43db138 : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cee51905628467baf18306bad63f459 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e857623c5ff3b993de60b3ee706d22d519a3e31d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-10 16:33:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4fa1663f05c68cd1518e220b2a600ecf005c337","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-12 14:44:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"7dfcc4d4c2d5c8658ef0ee42cad76c28a69f36c3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 11:26:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67b2ebb65b90ee8f1a75d18a6fcd33aa389ffb92","tag":"autogenerated:gerrit:restore","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 16:05:42.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b6675ab32f647afc332144f48785e305a00986e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:07:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/030de4c770b5468db647a4e33009f5cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d11cd9c85b5470bbc2e4512c48695c4 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/099466a26e3c417a877249774ea7dace : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e5e5f764154db8bb549be710bfd30e : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/082befeddffa43cd861458c66e218ffa : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e22a6b1f25f44c49417366869f35f88 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daa86a2f70344b9da721a2297c0db8f2 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6825f47b7bf74948acff232543ed2856 : SUCCESS in 4m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce45a819b3ef4412bedc332857c23967 : FAILURE in 4m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1edb85b718ef4476a23e03561f2af3fc : FAILURE in 5m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/367663a8ac674c91b1c8928b7e571bad : FAILURE in 6m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f5627dce820d46c6a4476f369c47b46d : FAILURE in 5m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3daed893304a4312aa1e28bb60ac9af3 : SUCCESS in 4m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6dbb60633cea45f1a2203a3443701ae3 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5f277d0919044638a7bbd3ad0abab741 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ba805353608d424380b030272d4c4f28 : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/47798c174618472b8d3057d9d669a236 : SUCCESS in 5m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cb7c02fc04b447b9b2882fcdeef1b744 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7324f1db4b954fc885adb8167be0d85f : SUCCESS in 7m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fe2a276a83324ec296d70586ce6639d4 : FAILURE in 2m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/804ef3ba945245ffb5faa9c5fe9c2c4f : FAILURE in 4m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/aa728ec660684bfe94fd657447e20d4d : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/72d1aaaddaaa4e6f9835ce2a86d66faf : SUCCESS in 10m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b462c910157f40a5a6600e85d30b76bc : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4694c783bb3648b6a5aa052804729c12 : SUCCESS in 11m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/42e3719a6f5549efb780a8ee875d2489 : SUCCESS in 4m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6fc9baa11fff4125ba6e8aed7b831011 : SUCCESS in 5m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/30d3a328782a4a39ac1d24bbda710e9d : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/041b321450fe4ac799e7f78c577ed73c : SUCCESS in 23m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7c04e8ddbcf047378ff75a76356b5b1a : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f359076c44ee4f9abec067cd2f5f24c5 : SUCCESS in 15m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1ceac84a02a649ebaadae214f3755891 : SUCCESS in 6m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/291db00865a9499a8b231e0b71bfb242 : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21124efd408447d08ea274b239b5a542 : SUCCESS in 28m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/758ec95d66bd4251bb1968bf160941a4 : SUCCESS in 3m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/582590a5678b45b599d2047e431f7276 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc5576ae33ce46f0a0e8b7de389a6ebb : SUCCESS in 3m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/64e53b581a47478fb980dc73b218ae5c : SUCCESS in 18m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0343511feff644d599f9c03a46d7bc2a : SUCCESS in 5m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8993bf6d487e4c79967e9d9f27945231 : SUCCESS in 47m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6e0693ff3574701bd8e289e3ab7489d : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4196b2386d10c86a70c04a0c0a8d379abcbd9cc9","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-30 07:33:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8c42ae235b7b5ecf2b259670c6610ba504fd271","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-30 09:33:46.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"004a2ceef1be507fb832633a931ecd6058d94389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 09:36:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba063978f91845109519ede5387f848a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7c4fce5a9b4859bbb632770fb5e11c : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb980dd8b524748a573001a60840fc3 : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc0d0f55c8dc4a02a8c531b18f6e6a2a : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/949c4becfa90480bb40f32daea942e46 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c73b7ab22744f46a09aaef87879049c : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cad3875ff7144679fcf5cad339ec607 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ac9e509d0ba4543885c3e0d1e8aa3ee : SUCCESS in 2m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dec7a35632354feab30217f8bfa9d838 : FAILURE in 4m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e51bd472e1194c88968febd38fda0b6e : FAILURE in 5m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fd65eac6a7304cbc8c28ff5c5f2ae9c6 : FAILURE in 3m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/84beb7af963a4b7da2cf9c832b83348b : FAILURE in 3m 27s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/43b7d776811d446cb14e7b1313ea91de : SUCCESS in 4m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/728110afee0142eea7890018b393e21a : SUCCESS in 3m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3f5e16e319fb4d53aa05cde2eeb32990 : SUCCESS in 3m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8a3a0e5a998a461393efd8b83390cb49 : SUCCESS in 11m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0846396063b34103a46b5014d13a77d7 : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d2e36e11c3224a268d22c3cd64bf5f0e : SUCCESS in 4m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a89d6eddc9cb455eae271ae2b06ab4e8 : SUCCESS in 6m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8f9a12bbbd834b95b5460d9353f0a5e0 : SUCCESS in 19m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b93a2d291d04ef7b3d9a04a51beafb1 : SUCCESS in 9m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/083ea075581449ea9bcf10d68d69489f : SUCCESS in 5m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd8d064f2d824879818dc5e0e9a828dc : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cee7b24fd6da4d73a905fb35b50b1a87 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ad4021bd1a684434857069aa0334b95c : SUCCESS in 11m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/96a93d814c1e4e0bbb1683a10842629b : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/26d46e095d384db882a12fcee6ae1ba4 : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/658780227cb646d5bb5d103e54ea4a6a : SUCCESS in 9m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3ffdba22fcdb4705a85833a37e7a99ba : SUCCESS in 19m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b370bfbd2191474caa9616b2fd68a9ba : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/520826bf408147949d976dc2338ac2a4 : SUCCESS in 16m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8889ce69cfb94cb3b2ef36797cd11ca1 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ae07e456152437b815e2ae27a0c978f : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f13450bfd604645b93a4592ad9935c7 : SUCCESS in 27m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/429358442c2e48f38452906a3ea51b5c : SUCCESS in 6m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ce38a67385cf4c4f99e54a93aad90fd7 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/493f2dddc327470b98b3f4fd38fe44f5 : SUCCESS in 4m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/24d343197c17422aa1e94f45322d0955 : SUCCESS in 19m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c5871994d92e42369544346e1f6f3537 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/280eaf00833b4a04b3cf9a17c873b586 : SUCCESS in 1h 08m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e41f286d70948888d3a2cf2fea26ba2 : SUCCESS in 1h 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c4df86fb5c01f8665943743f486e44a2e0dc5ba8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-30 10:48:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c3eb265683f7aff83b0fecff2cacfa3bef4675fb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 08:53:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c53fbc3e5c5880b32afc80930417cace9707344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:46:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e092d7aa1ad450794dad3f77b79b2f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c24f7221f84f4c819bd5745d9177e0 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851c4ce6878743a987928d61a8da421b : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db9f4d6564d348128ffe799052876ea1 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b46fcb2ac50a4b5497c225ee67b0e039 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6b8394f35af4c8c8911f759c48f34ec : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c682321fd4a4c9583d711f290b90bff : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e357bde9c1cc4aa4bd3e21d0831134b8 : SUCCESS in 3m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/068fc6ac180c47e283832d756047b418 : FAILURE in 7m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c5c1976500e94df7a57be56f36897058 : FAILURE in 3m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fb6056be2133415db4f126835f1a4c0b : FAILURE in 3m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d7c5c529c6174938a1278b8d9600a0bb : FAILURE in 3m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/530b3ac6a9b046e6a924c90c562b8048 : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/53ff20c2ea1340ec9548cf577eedec1c : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5eaa5df1c3034320b784ee346b69eeec : SUCCESS in 5m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2076bc824f1c4c2c860d1bd692cc6a08 : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/14eaacabe6164274b630bac68624da9c : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5e6e7530bf4549b7a0368ab2756b28af : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cddbb480a22e4b5d971190d4aee53301 : SUCCESS in 4m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6b39c46ba79f46768cc322ecf9be1800 : SUCCESS in 20m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a280f3eb41147b4aa0cc9cecb631885 : SUCCESS in 6m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a70246049ea4f5e8433886d6167a79b : SUCCESS in 8m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/94cc02b5fbfa46eeae01abbacd4f8018 : SUCCESS in 13m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bc2383f8ec474bf1adc1839daef66640 : SUCCESS in 3m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba5e227036aa498ea9108e579f738e90 : SUCCESS in 9m 33s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4165a2a6e7144cb39ee29b2db707d672 : SUCCESS in 6m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/893343325ad141288c0140eba623ef94 : SUCCESS in 6m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b725fafc58a04d1888ac414aaaa7e9b9 : SUCCESS in 8m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f78789fb9fab461aa8da60db7a5c38cb : SUCCESS in 24m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4902ca77ff554405b88b39d71a6f81fe : SUCCESS in 7m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/eeb5ec2bb16c4d7d902bf0b8bbf6a201 : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1538d8ae8f364a29b7c17c9df5a09a7b : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/738f94cec1c144ca9eba2eece099bbc5 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54ffacd1c1d648c89e02b4c57c4bdc6a : SUCCESS in 26m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3424eba2e73f440bb7a29ca1eed79192 : SUCCESS in 5m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3ed62b40198a40e3a6dc0b2b6a0dd191 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66d0b94fe4ab4c88ad94339033cbe037 : SUCCESS in 3m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0e5cd75af09c4b7c92cd2df51b19ccee : SUCCESS in 15m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3f661554db7a4c32b5ce1aa1989905ff : SUCCESS in 3m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/816de4093b9c448e980f32584a6dd841 : SUCCESS in 1h 03m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cefb02c0356c416e8bd3fc6415a3d827 : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8ef2f596818a6171ec8434cadfe93b18c8c8b1db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 15:45:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1dd7037c0e0140cb5e8179d68c4d4a50c01700b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-07 15:39:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ed2f45d06e128409e1ff40e16ecd5cf8fc0c2c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-07 15:39:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"64476af1a4ae2dafaafcf1d36e92006713d630df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 17:43:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f20abee1474b44fd9d422f563de4ef98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2d108e755d4efc9316f366ef332f08 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b74c315fc17434cb96d237d0ed437e8 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d0532de70174b6aa6fd525e00914c87 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e9d550b30ef4c2ab8b2f21b7469f4d1 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d61ae993edb643c68c2c33fe1a290544 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/783ed5d7d0c142cbb63e936fed30bc3d : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9489aa584a34f73a78f580a8807f147 : SUCCESS in 3m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/90ec56f84b66457282554cb2f805c527 : FAILURE in 2m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/920203decd3d4c1a864e5a0be86e8334 : FAILURE in 5m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fa55be1d14ca48cfac0da877f9e4b8f3 : FAILURE in 3m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/96a8f399070c4acbb8673dad005f6a7f : FAILURE in 3m 23s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7286beea1cba46b5a2c8c5ff478efe45 : SUCCESS in 4m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5a150758eab34f2e96d6c0230b19eb05 : SUCCESS in 3m 14s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ded98747e02d4f169dba6d3e0f52d2bd : SUCCESS in 4m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/295c01045b2749edbf42152b9b12b052 : SUCCESS in 7m 12s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1d281863d1b24f54b5d0a4e9e4b29487 : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9e38b13d32a349e9ad78fb5ec15e4c98 : SUCCESS in 3m 58s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/462e552ac2b341b396517ae45880a0eb : SUCCESS in 7m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e008c48ce3c84641a3bafc2a6606f52f : SUCCESS in 18m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f93d23cbd5a488681f6afddeb965e9e : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a3660b3962964ed39f15bb801cc948f0 : SUCCESS in 7m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/857b627a17414406942fcc3b0aa92d75 : SUCCESS in 9m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5ace7f21423a4bca821d718bc4d445ce : SUCCESS in 1m 38s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e899589300374a2d8fbf7879185b312d : SUCCESS in 9m 05s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e92943e198de466f83ebcf7b62448b04 : SUCCESS in 5m 11s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/a58d7a19dc7e44a6b265a5fe529d253a : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/c78b4e782766469aa7d4de38c84299b6 : SUCCESS in 8m 11s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ff5a53c5b96b45aea12d5ff177bc3855 : SUCCESS in 21m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/52ee0e13b21f408898e30db43e5fcee9 : FAILURE in 4m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d0544e08ee6948998146953ac48b2751 : SUCCESS in 13m 40s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3573dc7e64b3416fb1942513b74eb47a : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a72fb6a4b0846da85aad93505872b8f : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4443860ad7b44cc8bb7039181233fac8 : SUCCESS in 19m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/008482c9fc6c4ea0b9dfd2b80b5dc3bf : SUCCESS in 6m 51s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/cd59078277154d83a6fd603519edee4c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17a337cd51264592b5f5751f1df0ccdb : SUCCESS in 4m 55s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4b0e35da10f542bb91c3ecb9e5a514bf : SUCCESS in 7m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1dc63ba1f83341c4bd0b7aba3425c3fc : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/869a61f1ba844a58adfb6121b27ea8b8 : SUCCESS in 38m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32187b394360467888c891ee9528d9f2 : SUCCESS in 1h 58m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8adf966a004740ffe744a27e5c79e715ad639319","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-14 15:06:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"312bc1df3417bd1431dddc87037f246fc16b3a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 15:31:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a69e1aea80c4c74ac1b6062f790c58b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401bd6af125c4e67866178b6cc8b5d08 : FAILURE in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958df310879e4e46a5ab3b0657992946 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fccda02b7c054b8a91a5cd4c23c43ea2 : FAILURE in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31f9cb7dd6554740840043bc200ea92a : FAILURE in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2accdcb55e114495bd4d214c12a49b00 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2436eef7c374aa7a1a613f711b7ae3e : FAILURE in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/044ec750c7fe4d7a90f0e7a1c135d1ce : SUCCESS in 4m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f815f3c1f71436d8ef2d0a2b44f2af5 : FAILURE in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f193272c7e8f4c388c0b89a4666bc24f : FAILURE in 3m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0b0e2bc1160f476f897356c748195f82 : FAILURE in 4m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9584ac0072f14a2c9355daa40ebb2094 : FAILURE in 5m 01s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a7f3e9f738d84344aedbb69eebfb40c0 : FAILURE in 4m 08s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/1eb1f278aaa9420bacae5a037f044433 : FAILURE in 2m 30s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/6d125cd69a43486e80189b2951c4a2b9 : FAILURE in 2m 47s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/096e830e410b4d49a4db77313f9728bb : FAILURE in 4m 21s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fc480ecc056d4984abfbecba9103b207 : FAILURE in 2m 40s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/986be66eb29c4de9af3a25b36c468595 : FAILURE in 5m 16s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/07cb72d1f6254643829074eb5d79d76e : FAILURE in 3m 16s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cfee1c5ba1c444339f37b2983763e051 : FAILURE in 2m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63db58511b7d4c7ea10a846769e77d95 : FAILURE in 6m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a6348b7b302d449cb1a4b72e6f900b94 : FAILURE in 3m 04s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0eecb8b8f9bd4d7fb56dcadf9b3d9746 : FAILURE in 3m 05s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e0ae4c3c531a4b2f88d01442e5c9030a : SUCCESS in 4m 38s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ea4c534a1cd44115b2e4ece4cb96d6d2 : FAILURE in 3m 02s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d0a5da049432420aabe8a209fdfa1bf1 : FAILURE in 2m 47s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7adf4412a18343e0b8f19f6dd01d49be : FAILURE in 3m 44s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ba67e5229c0349d29038b9ece019a8cb : FAILURE in 4m 07s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4571b250f1bb49e7b5af9ab05aa26246 : FAILURE in 4m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e3fcb37c4534e2998241b31db0706be : FAILURE in 3m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fbc8541ad95c48e59df53d86cb62425a : FAILURE in 3m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/588dca30acc745d889eb83c705972c06 : FAILURE in 3m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edca3ad765c24bde8ce1c6e5173e9465 : FAILURE in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e51d015c2ae64038a82fb6ecff73465e : FAILURE in 2m 41s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5a6f56b71c9249298666d3a7645fec0c : FAILURE in 2m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ba4eeb4e5730473cb9513b8ef7afec4e : FAILURE in 2m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a6de1811d0d48a7b8f0eff87c50e8d0 : FAILURE in 5m 16s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/37de2de5d49a496fbb86e13ef5fc1e6f : FAILURE in 2m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7c996e7650ae4e8dbc57fe31db114e76 : FAILURE in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3590feac7b414d208bf45d3e55e11ffb : POST_FAILURE in 14m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b587be2f3f814c02807ca9551233bb2d : FAILURE in 5m 41s","accounts_in_message":[],"_revision_number":4},{"id":"29becc5c317d88ffa8cc11db96237ca2c49ecf6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-15 09:50:20.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c045ac3dcc405bd55d375bee35d42f57d19c0b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:45:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7354f12341214d8ab7ead08c0b6a27ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb222b1e593e4693b3f6ee428b26adf5 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf58a73c57e34c04956b18503c9628f7 : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/544567dbce824d68b0130f0edfbfe265 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/107f7908311645348674175b3b15d166 : SUCCESS in 2m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ae57fcd7ba545598eab2e09794b3092 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c24aed7b33548ee97465081e6d56fe7 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62ce018081074ee0aa42f07a2f33e845 : SUCCESS in 2m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef9aab5b463641b0ae835d4cc451c28b : FAILURE in 4m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/491c4b463c2a4ae5b74fc1de8e4b8095 : FAILURE in 5m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bcc815d7d316438fa5d104cfdd245cd7 : FAILURE in 3m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0fd71485fc21458f8b27f47f063f48d7 : FAILURE in 5m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2918e1af80104ab4a438b338fa1cae66 : SUCCESS in 4m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6d7af867c7b84aef871a99c90d8360b8 : SUCCESS in 6m 51s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a48762c6d5a34618a0e9becf1866f215 : SUCCESS in 5m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3ce3b3262e7d49e08eb1061956f6215c : SUCCESS in 10m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/994af40cf085482ebfd76531c6f37f53 : SUCCESS in 3m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a769fa29cb40421cb6f6f071a7ccae38 : SUCCESS in 5m 21s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ef5b206becd44b76bf2ec3f9247c598d : SUCCESS in 6m 00s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/70af940d82a4426d93f00df395acf2ec : SUCCESS in 17m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cfcef08ea464c0581a3d2e747476b88 : SUCCESS in 10m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/d0fee662b694479fa063cc29549ef824 : SUCCESS in 7m 30s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dd5cce056c654c82a8b80bbfac519adf : SUCCESS in 12m 29s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b49722ae09554d4fb6b49ce6f8481541 : SUCCESS in 1m 56s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f1af5aed239d4ad09bcf502791baa2ac : SUCCESS in 9m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/281dbf6f0811471b8646677daa87dca2 : SUCCESS in 6m 40s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/28a2001278c14acca1a357317271c523 : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/3c311336db2e44e1ae7f2c50dec151be : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/404580defc7448e68605b00d0ad520aa : SUCCESS in 18m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/716f4c3116ab471c9bed6123fae5e56b : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e4228c9f8091458ea9498a0597ee8f41 : SUCCESS in 14m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c82d3d0bc05b48b28aa45fdbd6ef6b34 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3c65c55bf9546c788613042112074cd : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f13698654b042a88eb8148bd124cb68 : SUCCESS in 26m 05s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/caf91a88e27a4145969a4b683a87c006 : SUCCESS in 5m 27s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/390e6d4adc7d444e83dc6eac00741975 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80321216902f4a6b81e6ad77b2d74297 : SUCCESS in 6m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4913db461f1d47dd8e8a241da7c48e89 : SUCCESS in 17m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9f1f390d2a2741749dd298800336c05f : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9def7ce3c75446d88c70ea5029e1e1a0 : SUCCESS in 37m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f2fed108b8449dfbb3ec0321de384b3 : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":5},{"id":"43f9104482bbe7a7390ff033336deb5e7c686f06","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-04-22 18:22:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b98df9a6732b2a3321ce66317437922a9fb95ff3","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-04-27 19:57:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5a1732f16680af09411036c6ce157373e9c1483e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 21:50:10.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b651d1f01cf341819f13f739db70e8db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e01f172aa74223a1f90a62e6e79560 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/207a89c2f15942d3abbfd2bbac99cc16 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b6d16512e2047e7862442e7efb1f862 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2beff088b0b64b2e97464f0bbd778934 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8fa69c5ca3144899fe5a65ff1fbf308 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc9c3660689941be9af2b2b18414411d : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edd1ce7b74af434d911db503aa16ca6e : SUCCESS in 2m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e232ecec98d406f842277606dcb4ee9 : FAILURE in 4m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8c4dbe04db1e47b5b6a3a5b2fea364cf : FAILURE in 11m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ba357c27347f4b70b4510f9a7a214d51 : FAILURE in 5m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2b81859140ed4fc4b2afb5945264155f : FAILURE in 5m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d5cdf5b3044c412ea8a086c8fdf0f0da : SUCCESS in 4m 29s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/092d89aafa1e4841ab620db369932221 : SUCCESS in 4m 53s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e383fd8300bf41f7bd61a5b8614100f4 : SUCCESS in 4m 02s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d78461806cdd419e881aca0f95ea2cdf : SUCCESS in 9m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1d8cde0ebb5e497e952dd93026135e98 : SUCCESS in 5m 13s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/44af10a70a79475aab48232f0b9ad513 : SUCCESS in 2m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/bc88fc56015548ae9d8dd760690c3958 : SUCCESS in 4m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b1c7185a9e0e412aba19206e2283ea12 : SUCCESS in 7m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c12ad4b83754495b35562f1c063fa86 : SUCCESS in 5m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/06006a69af5a44498776b9be31b253b4 : SUCCESS in 7m 53s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/be64092b738c4639a38f3e746bcda840 : SUCCESS in 9m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8b1e10e4c49641c4ad3fde1c815f6b8b : SUCCESS in 2m 22s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a379a43961df47df8f161d459fc2a357 : SUCCESS in 4m 27s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f62eaaadc121477c968024b9c8e8526f : SUCCESS in 4m 10s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/54a67e19746b4662a4363497409fb60f : SUCCESS in 4m 14s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/411ecd4012fc42ce92a094d5c7d6a7af : SUCCESS in 8m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/877e6ae092c04159acbfa4faae7adb1d : SUCCESS in 23m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb989c5517b14501a54708eaeecade98 : SUCCESS in 6m 02s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b14f8162fee74546bbdd143e60954df9 : SUCCESS in 12m 30s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7d7aca0efdfe45e98343866d5b5d156a : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a352b5d9b8b4b709cf1b8134474e8ee : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a19dd456ff254e4c9b8b1ea29664a1df : SUCCESS in 27m 00s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/57cd42c569944312bd0ae6d6736924bd : SUCCESS in 5m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a1ba8afcc3474013b2504b241e0d3023 : SUCCESS in 3m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2334a3e11e54a61ad25b0ecd7ddaa85 : SUCCESS in 6m 39s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f492fd48af3948c6a33e1a6e740fdbfb : SUCCESS in 12m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b7d938b5dcab4d139bf5ec47b3832709 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0abbc79ddfb418993e581e242d13169 : SUCCESS in 45m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5739320ce2b941119638720ed3af8ea9 : SUCCESS in 1h 49m 36s","accounts_in_message":[],"_revision_number":5},{"id":"429baf50881a3bea06e282b6a9e2ca5c42a487b7","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-04-28 01:16:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"89a1f70204e55f350ac17b1a0fb95933f779d8dd","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-30 13:31:23.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"4d6890f4aef52c3979f481f90d940994b51c1551","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-04-30 13:31:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2f40fc3fb6fb859234e8c7190452556f1d50ac43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 15:42:58.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb14c372292943fcb40f2fae2dd36f30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d94febf2c9e4a27ac8b44807c9c67be : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ba8723756c4e3fac641c5d633fddb6 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fccf9b7e36d46cab919119c1f4eb03f : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba4893a335504e438aacff4afe3153b4 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a72a14359ca45f993ca8287492eb7b2 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df9b8c806e0d4e7796b9733bc82a263e : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ea4ff35247740f6aed3357a1eb54797 : SUCCESS in 3m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/50c86243b92c4ed9be34f90bc26ce011 : FAILURE in 3m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/51b793c8750f43ee9111e9152b24143b : FAILURE in 3m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fef3b05d2d024c42a9e8bdf3195336f1 : FAILURE in 8m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c8f5a133187342f0844dadf1163daa8a : FAILURE in 3m 04s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/59924d9e815748b39a6fb6ea9f4d5149 : SUCCESS in 4m 37s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/edf0ead3fa994364b064609d437c5d1a : SUCCESS in 3m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/9831016d101c4856be11fc7a7113b7b6 : SUCCESS in 4m 53s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c88a50ef00284a25af76bed4744888b3 : SUCCESS in 11m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2ec2227f3aa8461daabfe6e46c31f034 : SUCCESS in 4m 55s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/442c48b7ef3047dc9f4f75341f689929 : SUCCESS in 3m 21s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0cf33ad1a8e040f3a64c9388084c22ac : SUCCESS in 6m 52s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e8c22130be8d4e4cbc7c0227187c09b5 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19d953f6665f43c4a01ecb467493c11b : SUCCESS in 7m 20s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/41cf2575bec246999e620acc9195dd1a : SUCCESS in 7m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3e77722a3b554786b4e922a00d1632c0 : SUCCESS in 10m 48s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1401e2fe09f94594bc8d97306777642b : SUCCESS in 3m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3a065930131144eeb7729fbdc9997c07 : SUCCESS in 7m 26s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8efadd7cb9564efc99d7f0897dc0b96a : SUCCESS in 5m 42s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/69e884a653a94e80b493f2bbef7fc0b2 : SUCCESS in 6m 02s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/8d2160fde41c4575b3a4f1df6a753e9c : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4d7d76538a494020849161db45df5d0f : SUCCESS in 16m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/06c603536ae543e98d91655a4ba4d236 : SUCCESS in 7m 18s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a5f3fa535be34722bea748acac0bf799 : SUCCESS in 13m 09s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/211e9e3edee94ba2bb760dd10f72d18d : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27003357d15e4fa1aa2335be4ac8fbf5 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e51998cb2244c23b7707ebe96aab7da : SUCCESS in 26m 49s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0fca5aa0a34943d3928511f8469d29bc : SUCCESS in 3m 35s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b3825ffb4ec64422836b78f8ab0ee48e : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aebdadd499fb4f2c9a5c88c7425ef72d : SUCCESS in 6m 19s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8316d789cfd64d24a865a1ed2189f7c0 : SUCCESS in 17m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5b49af806a824fa7861d9ddc4b35f791 : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fda05ace92a4065bb042db506f1b0a4 : SUCCESS in 1h 10m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8655deba368d466190c338df2e0d1a68 : SUCCESS in 2h 03m 33s","accounts_in_message":[],"_revision_number":5},{"id":"a1414451191f2b6bf2fa8cb711d355073d5f8635","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-04-30 19:38:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be93b0244935f4d299dadaa6c325a02641716931","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 12:47:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"823738ddc07350fa84882a34e29557155e767484","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-05-08 02:45:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45d8a3e295ab659d61c3bb599bb9b4f0e4dfe7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 04:35:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2432265fa28447f3ae3027deaf50bf90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a496f38b40b4707ae650ea94902075a : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff51470b5b264e429b751f05c02dfd9d : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c8e36e4d7854a75954f444cddc7019d : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca8f3c59af604cb7a403e666a2a87eaa : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5c94d37aff84138985300bfe597287d : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad66ea03e35a4e80958882b2cbd9b630 : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ee943e8196942b68f7a83cfdcdb2e4d : SUCCESS in 4m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b0221e6c0b0c49eb972509f9c00bf11c : SUCCESS in 4m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc81ae1227664321b789690cf73a594b : SUCCESS in 5m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/22cde9527ce64936888d2d2c1d1fd07e : SUCCESS in 10m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cc8eb14e8130411092838942d5745b87 : SUCCESS in 5m 21s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/0b294ce72d024705b78c78608b5406ed : SUCCESS in 6m 47s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ea87fb1e09b54bf18112795bd505dac3 : SUCCESS in 4m 59s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/59accbdfbe274c45ad943d0802101c4b : SUCCESS in 5m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5895d57d689d4d20a619eee13c21a81f : SUCCESS in 11m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fd1917a0062e4739bdeda7e6fcffb7a1 : SUCCESS in 6m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ab41143327574f3e89856bef3647133e : SUCCESS in 4m 03s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/123ffe885a734e9fbdaa3d77d28fdf4a : SUCCESS in 6m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/16c45fe390ff46c0a5142595776240fa : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8504573788c14152b21c38992bde470d : SUCCESS in 7m 36s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/309760c3dcb14cf6affcd56a515946ee : SUCCESS in 4m 44s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6d0e7624823247e9bcb7cdf7d0cf39ba : SUCCESS in 13m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/04c926ac1b6642c99b1992152b0028a8 : SUCCESS in 2m 47s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/545f155172fa4afdbefb832c052259a9 : SUCCESS in 5m 35s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c4596aef76f14fa78017cbc129d81c11 : SUCCESS in 6m 26s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/17779580904540d48e4c3e1a785ed5f7 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/27b4cba72f524245938b2634dc75efd0 : SUCCESS in 13m 04s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/46378cdfa44c4065ac5d917955d3d911 : SUCCESS in 24m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d50c3064421a43a9b5c5d9c7728011e0 : SUCCESS in 6m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/698958e5d83d43f4b496c324a6b82cce : SUCCESS in 13m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/242088dc0f8b467891e8c75fdf704469 : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95e21fc5ed9a40c384e9a8a9250833b1 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a452a696b08463abda72ae89685bfd5 : SUCCESS in 27m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/382849d48044451a967b7843e73c2f56 : SUCCESS in 5m 49s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/10569f8e13914f27a6d7245375917249 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35f7358c460d453bb6e7739190d980ed : SUCCESS in 7m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8d39665b12ea4a5ba2d46366505fb585 : SUCCESS in 10m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/982fe83aacca4b9db61399b97a934464 : SUCCESS in 7m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caf8cd2e81dc451fb8cffef9b7e37de0 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a74dea3700244b39cbb1620b9c06a63 : SUCCESS in 1h 36m 04s","accounts_in_message":[],"_revision_number":5},{"id":"85414d4a48e04d0ac7128c4fe3dc609ecfbea626","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-08 06:05:53.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"42c0cb57c832928d08c93d628ac8d5cc54a967f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-19 13:16:38.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f13813aa246cd402de63072be27f5bbeaacc2f94","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-19 13:16:56.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1eb5ec44112a98f015e79f2101f5e424b9db0d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:25:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08cb511338264ff4a686fb7e423f751d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba46e7482f604b8c9899a385c7d9d6db : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c6efc8cdb54f93bd4c9319c389aac2 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d29a426709444aa9473cb590538681f : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0131ed1b84a3457296432a511ebf0cc6 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5191cdbd307d49f3a19ecfedcdc8f960 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cccf0e0302d74cfe8ed09a79ef05187d : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82b09ac30ca64f0186b2b99c36bf9203 : SUCCESS in 3m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6286b7c0fa8847389cd7a62d93fd8a45 : SUCCESS in 2m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e9cfb7fafb24402f866d962a460c2047 : SUCCESS in 5m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/efe9450d46d0499f9dfbde8be158a110 : SUCCESS in 6m 43s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e4aecfa378fe48eca38fd7a3160f16e5 : SUCCESS in 3m 06s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a08285db477e4192a113cf089817c5e4 : SUCCESS in 4m 29s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e700820edaf24f33af6743b5b4f33c65 : SUCCESS in 5m 03s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0c860ef848814a5899be53c77726ede6 : SUCCESS in 11m 10s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fe58696ebe89424cb9210ac0c38bea2f : SUCCESS in 4m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3cb5ca87cde6486f9226433a279940d2 : SUCCESS in 3m 18s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/fd9e68ca6b9149aba92239addba270d6 : SUCCESS in 6m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1cf8ee2c107d425b95c588b4ab75c788 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c31a8496f3744d08fe2e0551cf21da2 : SUCCESS in 5m 59s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/48a722aa0df440b98a3952d6efe0ba85 : SUCCESS in 7m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/32be8e6365db4cbea186ab4ac4464d9a : SUCCESS in 11m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/420218d5169246119d566f71cf35bdda : SUCCESS in 2m 01s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c3957e7e3b514d3985bbdd5fbdaf5025 : SUCCESS in 9m 33s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/bd098348187847db936b8fe38b4e6ed1 : SUCCESS in 4m 09s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/6c20c9381ae84090902e00b07751c0a1 : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/445366388f4342b4ab35ad7a2949f977 : SUCCESS in 10m 05s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/08faff1b0896423596572de1442c0bcd : SUCCESS in 22m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98e4b451dd3249f18064117f214ac83c : SUCCESS in 7m 58s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/375cdd3381ce4663a44cfae602995848 : SUCCESS in 13m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/40237c266834445e96299cd96c2402cf : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/557c35cf56d4492db0958b8416321609 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c81871d41fb2402d853662b2a41e2eef : SUCCESS in 27m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/bbdd84d20e9c440cb2d61726c8e022c4 : SUCCESS in 6m 31s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7bd831038c644a58bc378f2acf58bf7b : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43a58b38a2eb43f497af191f2e6716ba : SUCCESS in 4m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7c64598686034e63b35654e6073e74f9 : SUCCESS in 13m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7636816bddb54770bc55573c8298c95c : SUCCESS in 3m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/396dccb719a4427eb265a42dea7898aa : SUCCESS in 1h 15m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be0227289762415ab7a3d8d0c8375642 : SUCCESS in 2h 00m 32s","accounts_in_message":[],"_revision_number":6},{"id":"ac42d434a58e8f2b7f8c9953841ba4dd661b5bed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-19 16:56:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"18278d68f61a6a3bc6eada3858dc60bddf775595","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:49:45.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1acdc957f53f7efd365b2acf8a4c043341c86878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 07:50:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dd4ee42a0fb9be507323a0a5aa19d80b4e4996a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:46:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abad93d6bf754cd3aec9cf8fa5e49c4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b887c05f9fc491198b1667d7fc20ea7 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8988b4d975b4a1da8b85648ed079837 : SUCCESS in 1m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ae2325f9d814e498d2671cf658d30d3 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d235e85d570449c68a26547903c1c109 : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fcd8e2104a0414c89ffd41b5bc7cd82 : SUCCESS in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82070510b319474a9768bd0165e4cbce : SUCCESS in 2m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f734fabc22dd40b7934e9001d8428cb9 : SUCCESS in 6m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0e6a911219c740f4826e31d6b4e91166 : SUCCESS in 6m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/585bcc64f9434f75a67764e11dc1e7ea : SUCCESS in 3m 45s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/568482ee8dbb43809952e8a36a12db39 : SUCCESS in 7m 24s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c788eb86a81242e7a098f074ae6413c0 : SUCCESS in 3m 09s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8b53585adad642f6b8a7a21a5a880149 : SUCCESS in 5m 01s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/82d1ccae959b4c27a39f7add5ac47d3b : SUCCESS in 5m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ac7daf6891334fb68ffc8f8ce84b6dfd : SUCCESS in 17m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b0330ceec6d461ba00d1f4610a2a4ae : SUCCESS in 9m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f81c8cd0c32347bd85ae4e380258470e : SUCCESS in 6m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ec759b038dcb4f76bde44cdf66f8201e : SUCCESS in 11m 49s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1175587dc92942d9876d56f94c4ac94f : SUCCESS in 3m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c9864cf5fca14930a76f66c2e0753161 : SUCCESS in 8m 22s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/03d41cb8f8b445a69740c48b22af34ca : SUCCESS in 18m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/96916cc2f36d4d4397571c6f2cccf727 : SUCCESS in 5m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/15d4082c178b4a2cb12ac0ff3c66c654 : SUCCESS in 15m 14s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/9be4ca160d984cb28391ab1b47c93fd3 : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/428a522b6b16427090d28767e3cfa0b3 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29a123b30c4c482dbd8c63a42a07f514 : SUCCESS in 20m 32s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/bb5979bc28f74186833acfbfaf1cacee : SUCCESS in 3m 33s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/766f902c39fd4eeebce6283801ada83b : SUCCESS in 3m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7be543776f524fe280b135382c515e93 : SUCCESS in 3m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d400e764c3b94eab97bf998bf5ef1c3e : SUCCESS in 19m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/85917b051bdc42f89b7358d6b0da65ac : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/625da698c3834bf6955470363479cde8 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72439f38391c4537b55306241b3de074 : SUCCESS in 1h 51m 17s","accounts_in_message":[],"_revision_number":6},{"id":"36f256628d810b054387487f5e966b0e3836c225","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:46:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"432f899a42129f6ace378dca058a21288f6fe531","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:47:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a47517e044d4adaaf4481aa1d3d93bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0db10e38a44163ac3a43a510b64022 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6ccd87a1245c899483d81d74bc8436f626a00d28","revisions":{"1d39847e4ce047b458a4c99828585e1c37b769ae":{"kind":"REWORK","_number":1,"created":"2026-03-10 15:00:38.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/1"}}},"commit":{"parents":[{"commit":"9de80aedebed1e83c8a35a0a94ace6efc381224f","subject":"update constraint for castellan to new release 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9de80aedebed1e83c8a35a0a94ace6efc381224f"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d39847e4ce047b458a4c99828585e1c37b769ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d39847e4ce047b458a4c99828585e1c37b769ae"}]},"branch":"refs/heads/master"},"577274c909dc24311ae99b966c61f639c23b70ef":{"kind":"REWORK","_number":2,"created":"2026-04-07 15:39:05.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/2"}}},"commit":{"parents":[{"commit":"a7a811d3bdb3c16e273dea824cf0fd5e1dad77f8","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7a811d3bdb3c16e273dea824cf0fd5e1dad77f8"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-04-07 15:39:05.000000000","tz":0},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/577274c909dc24311ae99b966c61f639c23b70ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/577274c909dc24311ae99b966c61f639c23b70ef"}]},"branch":"refs/heads/master","description":"Rebase"},"747f1549654796a3576c5a49493ca4302e2ce516":{"kind":"REWORK","_number":3,"created":"2026-04-07 15:39:56.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/3"}}},"commit":{"parents":[{"commit":"a7a811d3bdb3c16e273dea824cf0fd5e1dad77f8","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7a811d3bdb3c16e273dea824cf0fd5e1dad77f8"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-04-07 15:39:56.000000000","tz":0},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747f1549654796a3576c5a49493ca4302e2ce516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747f1549654796a3576c5a49493ca4302e2ce516"}]},"branch":"refs/heads/master"},"50714e0702db57f0f7309303b9395028a3a9b5c1":{"kind":"REWORK","_number":4,"created":"2026-04-14 15:06:03.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/4"}}},"commit":{"parents":[{"commit":"84e467a02fc484ca063ae383ab95d04f530090fb","subject":"update constraint for openstacksdk to new release 4.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-04-14 15:06:03.000000000","tz":0},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50714e0702db57f0f7309303b9395028a3a9b5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50714e0702db57f0f7309303b9395028a3a9b5c1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"747f1549654796a3576c5a49493ca4302e2ce516","theirs":"84e467a02fc484ca063ae383ab95d04f530090fb","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"9759b0b53743d98dde1cf2efce22ce812c28649d":{"kind":"REWORK","_number":5,"created":"2026-04-15 09:50:20.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/5"}}},"commit":{"parents":[{"commit":"84e467a02fc484ca063ae383ab95d04f530090fb","subject":"update constraint for openstacksdk to new release 4.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-04-14 15:06:42.000000000","tz":0},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9759b0b53743d98dde1cf2efce22ce812c28649d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9759b0b53743d98dde1cf2efce22ce812c28649d"}]},"branch":"refs/heads/master"},"6ccd87a1245c899483d81d74bc8436f626a00d28":{"kind":"REWORK","_number":6,"created":"2026-05-19 13:16:38.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/62/979862/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/979862/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/979862/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/979862/6"}}},"commit":{"parents":[{"commit":"84e467a02fc484ca063ae383ab95d04f530090fb","subject":"update constraint for openstacksdk to new release 4.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-10 14:57:34.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-19 13:16:28.000000000","tz":0},"subject":"Update Django to 5.2.12","message":"Update Django to 5.2.12\n\nDjango 4.2 is going out of security support in April 2026. This is\nthe last moment for us to switch to Django 5.2, which has security\nsupport for two more years.\n\nChange-Id: Ife9bfc0ff35c7b636eae1c474522f7f4f25781ae\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccd87a1245c899483d81d74bc8436f626a00d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccd87a1245c899483d81d74bc8436f626a00d28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
