)]}'
{"id":"openstack%2Frequirements~980271","triplet_id":"openstack%2Frequirements~stable%2F2024.2~If692753ae6e73bc84cd13459fd0af876f19fa94d","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"If692753ae6e73bc84cd13459fd0af876f19fa94d","subject":"update constraint for oslo.config to new release 9.6.1","status":"MERGED","created":"2026-03-12 12:38:38.000000000","updated":"2026-04-27 17:11:16.000000000","submitted":"2026-04-27 17:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980271-new-release","meta_rev_id":"b3a4c8d2e6eca3b48e1dcfde0fca1cbef3e4e5e8","_number":980271,"virtual_id_number":980271,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 17:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-03-12 13:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-04-27 14:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-04-27 15:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 13:17:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-03-12 14:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-23 05:30:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-27 12:13:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-04-27 14:18:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e3db3dc971d9636677d496da406f98b558bb7ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-12 12:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0790ac76928bd8182ff044085a4ad9b14b89952","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-12 13:17:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1861cd82a6ae25dc40362283b2046a0122565d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 14:27:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb03428e989941aab1b87adb84d1a7e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c7ddc0091d4969bd8c829363c79aa2 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd9f684d09eb4222a718ce77ac962e55 : FAILURE in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc1682b73950444da6a54364f8c514ab : FAILURE in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8b66fc15dcd405399b401de3f812bf6 : FAILURE in 3m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d606e94a91b2403eb5ccb4f3d72669c2 : FAILURE in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1aad7adf025940509086125fc32e2993 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2fb87cd0206d43bcbf7885a21f71402c : SUCCESS in 11m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df95044ebeab4d9380ff41c05027a513 : FAILURE in 24m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/37860c2b018f4691a6a70ab4e83c5e12 : FAILURE in 10m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/668c3624579c4d9595a7470a29c085d7 : FAILURE in 4m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/819f753925cd440ab56d623b871b0633 : FAILURE in 4m 45s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6db6a20267cc410dacb51d1c978cbb1a : FAILURE in 5m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ba5f93ea91964e27a2a6793cb5a49b9c : FAILURE in 4m 03s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1e5b4600adf34cd3921e03c8c2a9828a : FAILURE in 2m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4772fa40db9f49d69dfc6b12b6f657c3 : FAILURE in 5m 41s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/fe51cc325d85429b96af2c157aabd253 : FAILURE in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e8b2df2c57e04c2ba7878a93dbfc22c1 : FAILURE in 4m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8bb40211cba54a039cc295520a23eea7 : FAILURE in 5m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1df593be4854418d84897f4fea6ea6f4 : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c18d631ce8a84ebd977a111fd0df1ef5 : SUCCESS in 10m 37s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/66d3cf26d0cc4042a0910debd16184fa : SUCCESS in 7m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/07166a09ec7e402f986a8e8e2ee3dbd4 : FAILURE in 3m 12s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b1cf2ae82ea243d1b795318c7ec9ef4b : SUCCESS in 2m 29s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/e67817794c874cb2a52c5ace1c3ccf5c : SUCCESS in 7m 19s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/6c1388509b6a4a74b7cfdc6b3aa802ae : FAILURE in 2m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/30f2b8d98dd84d96b84538b4ab955973 : FAILURE in 4m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9609b5c85c48409daf13f2c1d259b87a : FAILURE in 4m 19s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9c68b148630b4abb8d3bd76cadee0a0f : SUCCESS in 18m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c6ac41849b3f4451a9392de7f0b5fd0d : SUCCESS in 7m 08s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/af4251c4a096474e9ace680f95b76972 : SUCCESS in 12m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0730e18f09274b9aac5597369012d28c : FAILURE in 3m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8bbaa27c23b434cb6581e3b933b8346 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e594c9b0eda45bba8a065c9dc63613a : SUCCESS in 23m 48s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/17ec3982370f44c99e55bb235fc813d4 : FAILURE in 4m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/687e3cee2bf34555a13c80b01dc64e22 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d44d197b126645fe9678c4cd34428580 : FAILURE in 3m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b0026697cf31456f9519b502bc019d60 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c8e506005704f3399df147e926e57c9 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6940ee672e924ca993429195329cd305 : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ba2bcbd4d2afd02766fd702f40a27bcc20d87ba8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-23 05:30:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab1a4e74c68a8dee34987aa90319b9753523c401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 06:53:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eac590f125344ce08868382098e467e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75811a7ce0684dedb5717128748d4fdd : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6180af4d96584d8ca2d8f07e9c1a7d5f : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1101634a60747f5a647306e808afc0c : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccdec2d9b0e6430aa096146e18c9758c : FAILURE in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8739ecdf44f42069a9c116d2c7a430d : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81051894bba2478fb7fb381b8af7d22c : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/febf43f1737e4ed4a5e08f89756f4c41 : FAILURE in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7110e530ac0048b9a138d3f3604def28 : FAILURE in 2m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a93bb2a5444842f7907e6d9521ad54e9 : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/50b927df458e494a8bf8a5c6897c7aee : SUCCESS in 17m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bcc2d5bb4e704ce9a7e227f4bbf4c53e : FAILURE in 10m 27s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9405102a0e44489a92ba7ba567a8659b : FAILURE in 4m 29s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/50326dc0515f4a42af3de2fa40607cfd : FAILURE in 3m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f31d57c46ba444c6bf8d096e8ebccb1c : FAILURE in 3m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ed3028053eaa4fe6817438941ec998dd : SUCCESS in 11m 00s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b137bc6f104c4e5db946fdc883e00f20 : FAILURE in 3m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cfbcd563a635474c962f497300f08edd : SUCCESS in 5m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cdb6f1ed9c0d4f9db1cd3b6e2c398953 : SUCCESS in 8m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e66b915f2f7f42498725ba897e202736 : SUCCESS in 13m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/024e39cb4b444f9c9b88b0d3bca057bf : SUCCESS in 7m 52s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bc60684b705142d19277a7471e8e9262 : SUCCESS in 7m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c09313c4db3042c4b2c8b223c6e76758 : SUCCESS in 10m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/11df322af5de4d9babd1477df6e53cda : SUCCESS in 2m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7c416988dade4bc78018191836a4d5cd : SUCCESS in 8m 51s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ed76e771974d4bc8a715edac52623573 : FAILURE in 4m 08s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9f41bb427a0440ac98d1152f532b82d7 : FAILURE in 3m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/91728324b0db4146a2ce3b4b89252873 : FAILURE in 3m 54s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f6b0dd62429f49cb8688263ff5fee250 : SUCCESS in 20m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/87c361676ee64825b552c58776a6207c : SUCCESS in 5m 46s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c8f677baa8964a7b893a7191cc69a7e3 : SUCCESS in 11m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c94e3048e2cb4ff480a466b0d7880fa5 : SUCCESS in 7m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b9f3513e0984d4e928c2376af98485d : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0d8faf21bd14e35a7bd0bf6581ce684 : SUCCESS in 23m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/95c8521391084fc7862e1b8b3ec83936 : SUCCESS in 4m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/930cd5a6992945e189a502d01f9ee5b4 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/514fc2ad617443bf99d4a3fda882c37c : FAILURE in 3m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7cecdb06b6ff484f9d34256c54792a43 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/618578f1318a4547b0814917a3e42282 : SUCCESS in 42m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cc67b6385d340979d86ead73464b415 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"dd784f3e884221beab2e2c61779402042747ff03","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-27 12:13:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"761afb6d695ace8eb741fb4a1fe63f2145333054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:37:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f259f4fc90454f31b820483959e8c00a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c18a4206284420aba4dc4ae1ba28ca : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a755e25d58b4bd0b042d27480395efc : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6088f7c5651e488991dae21f5a243954 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68cb5a36201345bc877632cdf97460f0 : FAILURE in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/492964cfdf264e22be2426f8a195e0a9 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c1de1faa6664e06a842d4112e1328e9 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82ad229d198e4809bfe35d84c147edbe : SUCCESS in 12m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/32f77f5e105243428bae6655316b0fc6 : SUCCESS in 14m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3d75460dbf5f48c49e91195f5b08b294 : SUCCESS in 10m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d3f85e5b6b7452a9c99bd455b7ce112 : SUCCESS in 16m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e298c07592eb436c9a66b0412c7fa970 : FAILURE in 19m 29s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5750315073454a669ae7aa7eb1bbf750 : FAILURE in 6m 08s (non-voting)\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/75ab8306732641edaacd4f4cd3d1449d : FAILURE in 4m 34s (non-voting)\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1abcb9d3d18249d297e9ee2d1c763803 : FAILURE in 4m 04s (non-voting)\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/edb8b724a3bb4e29a8f5a254c659b27c : SUCCESS in 11m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5283352a2cb44eeeba0823a8a2c8eff2 : FAILURE in 4m 32s (non-voting)\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f7b8d110aab0408e96428f2f7022f5f5 : SUCCESS in 5m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6d47096b12b04703beae4b89bb053a63 : SUCCESS in 7m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9bf7fcd6ec7a43ed8c49ec31945c2e2f : SUCCESS in 7m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9937b32830b84c0daa57c4f0510d49bb : SUCCESS in 7m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b214e70a120142ddb0da9222a6e92402 : SUCCESS in 10m 13s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a29a8e31044140ac83d71c4831de1242 : SUCCESS in 11m 22s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/93e972fb42394d3f949f3ce535a45ebf : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/da3ab205214644138a920afae3caab84 : SUCCESS in 9m 47s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3437d2b5e0a44516b69b24e790ac5062 : FAILURE in 4m 27s (non-voting)\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/367e2701653f49c3b53ffc420dd9f8c7 : FAILURE in 4m 40s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/57cb34a1baaf43f4a518cd295e264157 : FAILURE in 4m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/be75db46271c47b38c58c057a7c3825a : SUCCESS in 21m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d4371528af0f4a09a5fb13b3f9e21ca3 : SUCCESS in 6m 43s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4fe7a4648c5d4144951101e852e8de19 : SUCCESS in 12m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/55f5261f48334399ad218f51d50b198a : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1dd58f1e55d74d77911e41e60af37959 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f3acea38b6c463182eb5a2120d7e320 : SUCCESS in 22m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ed3fe438bea444ebaccd64778ba31f4c : SUCCESS in 2m 56s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/fa2b983890f24b9fb4e2a1467547fdc1 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8fdcfb4ad0247b1a76a93754c906e21 : SUCCESS in 5m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/91b478ca961d4033aa24aeb7eaaaea41 : SUCCESS in 19m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cc0aa05e5a44ba39d2b28f170e37c5d : SUCCESS in 57m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74ec396e9cbb47cca0f7dda077db9636 : FAILURE in 45m 45s","accounts_in_message":[],"_revision_number":1},{"id":"73b2801263a00b00de0edb62a9e3071adf6dfcb9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-27 13:52:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0899a555368d6f526e69b215538eef6b9104a11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 14:18:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a423cc793d42f4a5d35eca60274f02b330565429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:30:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0216ee62e64a03b785976acde97d37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efb81c6a28cb494abcb8f5568ed75479 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2de3c4d31f24acea32b3e12c1251e47 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/772647a13ee5487286157c137efe7031 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc31bbd7941d4ff3ab51bee397c2b73c : FAILURE in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f14f562d258480681ac31f19ed65f11 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d91b7d822654dc9b1e1ae18ee87898d : SUCCESS in 2m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a9316f8f37b49b9aa5d4d3c7c9e0b34 : SUCCESS in 14m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/982f06426ad54c46a7942c6750cf9bcd : SUCCESS in 6m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b4ced5dee4e24a60adf876c6756c0342 : SUCCESS in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0f5ab990c71b423d8e8c8aad81314729 : SUCCESS in 10m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/522a00eb295449219ebdbd164436535f : FAILURE in 26m 02s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b094461af51e4b35a5eb45bd01a8bcea : FAILURE in 7m 25s (non-voting)\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d47657bb541043cf8cf30199fa44a95a : FAILURE in 3m 50s (non-voting)\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f49decb126dc4dd4be1b9247fefff1f4 : FAILURE in 3m 29s (non-voting)\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b961533a83c04003881dd0f0adba67fd : SUCCESS in 9m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/da232c09b61447299d1b3f7d72bf968f : FAILURE in 3m 01s (non-voting)\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b140eb068fcd4bb6a5b88a82a713d350 : SUCCESS in 6m 40s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e0165a4ee42d423ebaee3401c155535a : SUCCESS in 7m 39s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f359d1b857ed4ea38873521a6583cbea : SUCCESS in 13m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99a3db7c8eb8465c83fd8caaae5fc6dc : SUCCESS in 7m 49s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/99ca3397a4204487a07d686821bd4558 : SUCCESS in 8m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e6b52cb5035f432d8c13cdd59d94a01c : SUCCESS in 10m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/dd070ab55969497195ebbc94cab35183 : SUCCESS in 1m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/de83a82e354a407cafd3b521038f44fa : SUCCESS in 8m 29s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/15291dc1ca144d6cb1de44aa4fff4adc : FAILURE in 3m 07s (non-voting)\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fa060f3c8c5c4e0cbc53712d6169f8e2 : FAILURE in 3m 03s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/721332a2c0bd44de8d787ef5e06269b8 : FAILURE in 4m 07s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/7c6da9e5edef43d4a4f8af1b02ce1651 : SUCCESS in 20m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cf899d7e0275443aa0d417476dcf6dfc : SUCCESS in 4m 02s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4fef8363f67c401cae4965d3a7ceef3a : SUCCESS in 12m 04s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c3d8b18d944643fd8e19f1cdc2fd1607 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/577ebc0840504104bac39b73d3744ac1 : SUCCESS in 2m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82302db1ee9f4b388352aedc159588cd : SUCCESS in 24m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/99ab7a5365324cd1ace6a005980e149f : SUCCESS in 4m 32s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/339b4ef6a6e34b72a3c14666d6708747 : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2b810a5e5ec49ffbef34dda09252f76 : SUCCESS in 4m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/24e574526dd04218ad67c1a56a56704c : SUCCESS in 19m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b3e521c98874b81b4ffb8a970dadd40 : SUCCESS in 1h 01m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca7ec0de06ae41c1abdfe9a4f7842c4b : SUCCESS in 1h 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0b3522cab0f2c1bbb30700224c0b4eb2bdeb5303","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 15:31:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c14e6c15470f40dfd7f2a432c100d15e5bf89742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:32:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"669e52c12048bfb1a4a0c054b76ce34b7bf54afe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:11:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a64f099d46f45f8ba7254e134545ad4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc9ef67944248b1ae7c76fb652aaf6f : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/192566ff242147a3a0b632e7ee50c66e : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7191b9bdb3ce4ebf9d594c88c015688a : SUCCESS in 2m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e6c994d8a204595aa353fc2aeed8338 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51d916fd5df14b72a202b7d5770a3cce : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f6f2b719e1042fa85319dfb68559e71 : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad063a02e9074bd886409832195010d2 : SUCCESS in 6m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1b3f1073b1b42dd8863de254a30d209 : SUCCESS in 10m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5cabb2a85da1482683f20e9918d789fb : SUCCESS in 7m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e0cc542d1ce9435695188913874c2ca2 : SUCCESS in 11m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7e00c1747f73421298b483efe12774e9 : FAILURE in 3m 56s (non-voting)\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bc52cd10e824412eaf488f4f192675fd : SUCCESS in 3m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b811df8a0b5c4b72a16c3aac8894a535 : SUCCESS in 7m 27s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f3656d496dec4dcc85c84505888503c0 : SUCCESS in 14m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8be49401b7d2402d8c36ceb3107fe431 : SUCCESS in 7m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7476fe5be94741dca5b170a6d2d410b2 : SUCCESS in 6m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e27e9d6a1520449fb49865235695d6bb : SUCCESS in 8m 58s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/01c89a6730c74d57a4ceaeb2b9af83cd : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ef26003c96a44e3c9937753e0ddb4de8 : SUCCESS in 6m 13s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/472be2b7ea1e4a08ae3a6ac7370c2d46 : SUCCESS in 19m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb213a59bc57470e93ed94d4a1ec936b : SUCCESS in 6m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f05d3c7c6e4c44c39fd86151a3ce2d50 : SUCCESS in 8m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/db0c3dfdfc6e494095e461f7b0d1b21b : SUCCESS in 6m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/482b6ec07e734b6f8424f0e53b2286e5 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/322d9e2afc3e4f6099b292c416cc1404 : SUCCESS in 25m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2e28064cdd1a480c9794a1f9bd4aedaf : SUCCESS in 3m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2c6a34aac3b5430cb49b60fbc728c52e : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e71c4bb14da45958b3bd0419e428359 : SUCCESS in 5m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0fc7a2fbedf243b889a1739d728743ea : SUCCESS in 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3cf13307dc24456895f7074f7c26c02 : SUCCESS in 51m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3442ef6f72a4caabc9c2fd900bc6e03 : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b3a4c8d2e6eca3b48e1dcfde0fca1cbef3e4e5e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:11:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80fff2e6d1044af7591cdc1e65d4d9a5114b4be8","revisions":{"80fff2e6d1044af7591cdc1e65d4d9a5114b4be8":{"kind":"REWORK","_number":1,"created":"2026-03-12 12:38:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/980271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/980271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/980271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/980271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/980271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/980271/1"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-12 12:28:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-12 12:28:37.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.6.1","message":"update constraint for oslo.config to new release 9.6.1\n\nmeta: version: 9.6.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I55c569cc07769af5b542c2eb15af1f4abb7b71af\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If692753ae6e73bc84cd13459fd0af876f19fa94d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80fff2e6d1044af7591cdc1e65d4d9a5114b4be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80fff2e6d1044af7591cdc1e65d4d9a5114b4be8"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
