)]}'
{"id":"openstack%2Frequirements~980273","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Ic4bdff8bd22295acee36815cff07d733b1cb1555","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"Ic4bdff8bd22295acee36815cff07d733b1cb1555","subject":"update constraint for oslo.messaging to new release 14.9.3","status":"MERGED","created":"2026-03-12 12:50:11.000000000","updated":"2026-04-27 15:13:07.000000000","submitted":"2026-04-27 15:13:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980273-new-release","meta_rev_id":"b93ff5381ff246845e311effd34e3c84e5c85202","_number":980273,"virtual_id_number":980273,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 15:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-03-12 13:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-04-27 13:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-04-27 13:50:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 13:17:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-03-12 14:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-23 05:30:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-27 12:13:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-04-27 13:50:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ade729d11d2e5e49b688eebd73116992eab48a55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-12 12:50:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3536af6a9a08a9059d673c1dabddce02f7310b35","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-12 13:17:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"837cdd1d7f5cfcacebbffd72633fda385f1e9857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 14:21:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdfe8a9361354921b19c948b6eb92078\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8654ef921b4a49a2fc3c380a423545 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/590cfe68218346c782a03d7f0b47edb4 : FAILURE in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43484e85cd7041eca0eee3e02f85aa4f : FAILURE in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b43c782b0bdf4d598362315b62b8057a : FAILURE in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/479e0686cff44174a1132e43cd6d7f19 : FAILURE in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff5d707f58ef4984a838f9952ee10f8e : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1bc6bbd23d504489b75326e2d5a4bde1 : SUCCESS in 7m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/97a46b56bfde4ce59a4f558af741ddf1 : FAILURE in 13m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da73ff3f007645438c6cfba4451b8eae : FAILURE in 9m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6fb17efb692c4c2fb1996764b2545e1f : FAILURE in 3m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3b2818584d874b388b0018cd71709c39 : FAILURE in 3m 48s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2a2097f3c5b746a797b6c07f1a64b4d5 : FAILURE in 4m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b06edee1be59497aae4dc73da7229cfd : FAILURE in 3m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/83f42a5bade041a5989833f5b0205f04 : FAILURE in 3m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8c92197e7cc54df0b0bcb1d96bf2818e : FAILURE in 5m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8855f8258a4547bcb682d1d25ae4354f : FAILURE in 3m 44s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c1d25afe8f9847658eb43ca74d3ba2d2 : FAILURE in 4m 22s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/700a5d25e1e644d199cbc3c465ec3413 : FAILURE in 3m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/22b4be09bdb44b0ab8de3c5fda37a03c : SUCCESS in 6m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/525971d76b594c92905d5a8040f1678f : SUCCESS in 7m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0bd0c2476cc54aab9aab1fe25b116593 : SUCCESS in 7m 11s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ebef1085b03048e59291eb483a819103 : FAILURE in 5m 01s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/08938df5fe244af7896cc23e27639762 : SUCCESS in 2m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0d76376d7dc446d4aa742455ee76be41 : SUCCESS in 6m 36s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/74745315e4944b93b7e6a592f97f2af9 : FAILURE in 5m 41s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e7dfac217f57404a88059768d107f769 : FAILURE in 3m 47s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a600f46a6194480d94c32db6abca002c : FAILURE in 4m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d53082986cd940f4b11f1137832ba610 : SUCCESS in 22m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2630e838a4454f3d9587aeae977a3a1f : SUCCESS in 7m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/02471c0f189146a0bb47ff0ef60b2f56 : SUCCESS in 8m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/bf32c88c7805418491546e63dd31d626 : FAILURE in 3m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/001bb1d0449446398d1efc13f6ae5eff : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce3e8c86baff4e32a4d491cf93a2353b : SUCCESS in 23m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/47d48db5977e430dbbddbc082b72a407 : FAILURE in 4m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/34c2e04e66834300abcf368189e1e043 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f83d259c36e647f58e8996d82539470f : FAILURE in 3m 31s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5ca778a590f142f6b61892d3b6e5ae48 : SUCCESS in 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a6976d235074277a05396e3587d481c : SUCCESS in 1h 03m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c89e8deb62a4bc88b50861c71addbcb : SUCCESS in 1h 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6d880a95e303978e6b35c605c51521b7718665ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-23 05:30:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"144c94a8d153d22400e3ed70f46b95f397a5d9a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 06:25:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9a688f79057403a84b2ab081e2e29a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d885b49ed5974651a47b46bd5014a674 : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00d9af4422a04783aa8b50706f01d08a : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fe8718f9869487cb06b234079deea53 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/019a1f4e25ce4f63a64d22451db63341 : FAILURE in 2m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bc7d2ddd3394832ba075a6d446f4698 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c22d182299a404489757f43b8e0b52f : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8820346cbf9840e8a38d7d400401e0e1 : FAILURE in 2m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/49d761af856f4b459e0ae37ec1117577 : FAILURE in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5a17a591f38840b9973822dbda1bb525 : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/116d3057c793413d942ee738d75ae6e1 : SUCCESS in 8m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a5820bf7b6d647d0a4612b31b77b63fe : FAILURE in 8m 50s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6ec6c86e8c2d4b7d99c7dadcb197827a : FAILURE in 5m 01s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6093ac2075114a69b6340aa51bd83781 : FAILURE in 3m 47s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/aa909daf75574c85a24ffb756db279ed : FAILURE in 3m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/421f4a1279504e2cbcb93d7f57ee8d63 : SUCCESS in 11m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5aa0291cb47041df9da5e65c23cc75c6 : FAILURE in 2m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/aaa3b6e385f24a8b868b7d52abf74e03 : SUCCESS in 5m 33s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2c0dccf2d3514d37895c9f48b9fd4ea9 : SUCCESS in 6m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/12b542ccdaee4849a836fa8f4057d602 : SUCCESS in 5m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12f4a08961f54d77ab40bff6ae796fe9 : SUCCESS in 6m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fcc6942fc42c487aa49ab6a28359dfa4 : SUCCESS in 9m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1c39d52216f749ac89d116dcf5b0fb44 : SUCCESS in 10m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0b3ec8ac063046f999b72d46dfff89bc : SUCCESS in 1m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a55da324186241759fb5416748449dfc : SUCCESS in 7m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a73e5b04828b4d9b97cd582cb3b3c600 : FAILURE in 2m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/62707827793e4b619d6ecfb5d7e94bc1 : FAILURE in 2m 40s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/14002fc80d6246b5bdfd07ed7deddf03 : FAILURE in 4m 28s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/71fbf925787948659840c2a7e7ec837e : SUCCESS in 20m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a06ed04eab714e33a3a536cc8a5e7faa : SUCCESS in 8m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b9cf59e2a6974f36a0d355852f710dfd : SUCCESS in 7m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8efd345d9aa84809bb50ce15a907176e : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1357a3729fe45fd9a09771bdb4b79f1 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5593f3656c674844b67ba2ad4a67a443 : SUCCESS in 23m 09s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/82c8c4474d5c4071b19aa2e3fac81b4b : SUCCESS in 4m 11s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c0f5f618f7014c01bf8d7cfeeefd6800 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31509ba564964b7c82880e90e70f7034 : FAILURE in 2m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/64f0453289be48439f37ee774129da5d : SUCCESS in 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72ee1acda3684a04b6592cf44d3fbfea : FAILURE in 32m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54635812ea75438e80006a0612f1762d : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c63cd02978211a90d3713fb11accb30d686e8e53","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-27 12:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"789b4391820971e59cdeb84d5bdbf252cf7d23c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3236b05e8fe24ea4b944c53acc5b517d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da9bb26e5ce472fb2cc77c86bbbd137 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28479d9185224784bdac03b83e75a8b3 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/184f685f216a4ec782d814eae0580d92 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b13e1408dc314f878f1dad62fb20e403 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc0f8c474d144c80a452d7c985dcd6e9 : SUCCESS in 2m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0dff8644d1b248d182c45aef8a928a4c : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/07da62a95da946e49f558e5116afcaa4 : SUCCESS in 7m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3805dd1ba5034c4fac2ed32627890b0b : SUCCESS in 16m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/509b1a65e7584712a8ed6921f6c97c7a : SUCCESS in 5m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2350dceb9bad407ea446490b34e63846 : SUCCESS in 9m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/26b65b9f48954e0daf32e9077d1b5076 : FAILURE in 10m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3238cb207bfc4c4cac7ab8b5a8861796 : FAILURE in 7m 33s (non-voting)\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/9726ebdf051b4268bac26f05b94f0224 : FAILURE in 4m 20s (non-voting)\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/48ac4c7a3f4f4c0892829ee0513250f2 : FAILURE in 2m 25s (non-voting)\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f9d345de23564a77ab28e975075d57b5 : SUCCESS in 12m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1ce57f1463ce46f79cb4d67982fc9a4c : FAILURE in 3m 53s (non-voting)\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5f4b7b0862184a578bba2b282dd937f3 : SUCCESS in 7m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9a3d6ec8510746729ed2668c5ba0eb2a : SUCCESS in 7m 53s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/14d312f3ed814fbfac75ba951ab2e655 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e18a8c27c51e47abaabdbb63ac252f08 : SUCCESS in 7m 44s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/52921a39188c4817b5e750317e298708 : SUCCESS in 8m 24s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e3f1fa73fcfa4046a08d3dd19ffd4b45 : SUCCESS in 10m 01s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/88d992fb51f445f4a51cc56a0be671dd : SUCCESS in 2m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/783a08f55fd840b78b63d9ec6694637b : SUCCESS in 9m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/1481c4a5007f4cbaa47d551ed65e954d : FAILURE in 6m 33s (non-voting)\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a7793134630c4ef798ee926924649c36 : FAILURE in 3m 01s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/498d0f4f583d49b3b51871d6a6293c61 : FAILURE in 7m 04s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/fc1a0aa1f7c2450a9774d5247276c7cd : SUCCESS in 20m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9c9492516f9a412ab7602863dffe9135 : SUCCESS in 5m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ed41efd9a1774a258c0df31288814f26 : SUCCESS in 11m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/74c812a2690d418ebad5640f5fd55270 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c911f7c74dc48598208b582a0cd32c3 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f08d78532608471d958988dd6f03b0d3 : SUCCESS in 23m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/96c3b293d8f241a29716cf458147f1f2 : SUCCESS in 4m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ef20b56e9074433388adea68521bc7b8 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3784203b162947e4abf82fa223ef7fef : SUCCESS in 7m 04s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/760f54d1fb2543c98c62be543e42f7ca : SUCCESS in 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac29ac2571f44946b37cfca7b187daa8 : SUCCESS in 34m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbcf214d979d43c9a109ea6a1bf0d04a : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cbe20cbd0199089d387e596efbf8eeb660e877aa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-27 13:50:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"824cf2e6f96f5af33ca327f9d1cb01ef3c66eb10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:51:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fafe1be3e42cbf47b6402aa9a34557afc1f886c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:13:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7bcfeee070499c8e3c8a22c6b9e80d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417aaeeffd5345d4b1df6f2c94c0c749 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fbdec0ef5714c3783fa3fac475287f4 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9a2d872927c4de6b48cd6b334f0c4a3 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85299245db6445cba00390074522544c : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5e914cb364544f5951d104c5b6f6b22 : SUCCESS in 4m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b6adc195d8943e2915340e3efcd3772 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9a302317e6a44a8bb0e9688e5c3fb0cc : SUCCESS in 11m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2b5205f6698d4456a6c4ef5d88d2c0b2 : SUCCESS in 10m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/129d38593a2d4de58fe4bfc44b55ce16 : SUCCESS in 17m 09s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/240d2b63a92744b2b7084323eb34cf4a : SUCCESS in 13m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/74197b5a689b425fad32dd1b7f85b0d8 : FAILURE in 3m 42s (non-voting)\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f26c40b49c724459b21bb7df8f7b4aee : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2d713795d8cc4425ae8370241fe71771 : SUCCESS in 8m 16s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cea20d1e54b048beb68be4d73074b454 : SUCCESS in 15m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cabf794681e748c7a7805b7601e50388 : SUCCESS in 7m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/684537e581364889bd9032bba7fa6e13 : SUCCESS in 8m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2031813f956949e684092f1e885f9e83 : SUCCESS in 9m 28s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/aecd7c2d17874ad09073775073e38681 : SUCCESS in 2m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/39c9d40f23744b7a860faef811b39681 : SUCCESS in 10m 06s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d74eded6c4b14e97bb3153381bf84a80 : SUCCESS in 15m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7fffad7ccf474163bf82dd4420527a19 : SUCCESS in 4m 46s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2ab43d4214c44f2f9960b817fdc6a1d4 : SUCCESS in 12m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/07c05c8c400d4882b23e9bae617c512d : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/982f277d87ca412da85d489c21504c6a : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/685882a258f1466a8177d51ff697696b : SUCCESS in 25m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dbde8dcaf6b3442fb4b02825609b04e8 : SUCCESS in 5m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1633f989c2664f218a49f1bf6e0ec06e : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59393fd97030485da717e82ef8ad0dee : SUCCESS in 3m 22s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/912ca6d036c049fb835b84c9992f31fd : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baf1533a9c834f34bf12da5cd257ad42 : SUCCESS in 57m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc5cece7806b4a41bd791f4997d97eb1 : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b93ff5381ff246845e311effd34e3c84e5c85202","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:13:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fd85a4117a6acdbe4c761de0796da9beced3903","revisions":{"6fd85a4117a6acdbe4c761de0796da9beced3903":{"kind":"REWORK","_number":1,"created":"2026-03-12 12:50:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/980273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/980273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/980273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/980273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/980273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/980273/1"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-12 12:40:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-12 12:40:09.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.9.3","message":"update constraint for oslo.messaging to new release 14.9.3\n\nmeta: version: 14.9.3\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I085f8b5a4bc31d89575524738484b7a57040d0d0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: Ic4bdff8bd22295acee36815cff07d733b1cb1555\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fd85a4117a6acdbe4c761de0796da9beced3903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fd85a4117a6acdbe4c761de0796da9beced3903"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
