)]}'
{"id":"openstack%2Frequirements~980479","triplet_id":"openstack%2Frequirements~master~I8543c244f325d6149d4701db13e1f6d23b0d5796","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8543c244f325d6149d4701db13e1f6d23b0d5796","subject":"update constraint for horizon to new release 25.7.0","status":"MERGED","created":"2026-03-13 15:54:17.000000000","updated":"2026-03-20 08:28:05.000000000","submitted":"2026-03-20 08:25:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980479-new-release","meta_rev_id":"08109e65d7a06a08fb2633a70cf401540e85f35b","_number":980479,"virtual_id_number":980479,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 08:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-03-20 06:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-03-20 06:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-03-20 06:38:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 17:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 06:34:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-20 06:38:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"505dfb07004269129cf030025c5ad8e9dee77f5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-13 15:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20f5758bdd1fc869a2e797b3f1da651cac5fcdb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 17:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ca66894e1143e9856fc6e7b9510611\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f43597dc84e4cec95f315be452f336c : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53f403955fc461da65c2f0143f252e4 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbed2e01d070491194d70513b22cdf94 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb5061e9cc9646bb8e1809225577dbf7 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b47e88cfc2f4647ad49000e2b685903 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c4b900c13e74c3d8bd596b337a7f9db : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/706c470e42724aeeb93d07f313b9cd5b : SUCCESS in 3m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c23cff1bdfd49bca9f786db4ed3a172 : SUCCESS in 11m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b7e3f21eee504e2dbc77839ca255fa11 : SUCCESS in 10m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/03f2a50291c24137ab2d9d74d0a54d6a : SUCCESS in 5m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e150f1aee3f048a09af9febc0e26b026 : SUCCESS in 5m 07s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4c2f3b812a2644f6a0088c8f5def9bd4 : SUCCESS in 4m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/df65382ff91548738a2f3b8f5f71147b : SUCCESS in 3m 08s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/430f48d3112a46cba11334ffc4fc7e91 : SUCCESS in 5m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/51b647e82c9b4945a77360d068607d6b : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4e2df4fec5cd4ab89c22b374de7491bf : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/123a2df1f1f34510add6fd7e963ac0be : SUCCESS in 3m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bbe73d8006b8443d894a2c9165a3b4bd : SUCCESS in 4m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/897a284afe924d8599b09b4cdbb615c4 : SUCCESS in 18m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e269fd7d76794e198c2da6443699453f : SUCCESS in 9m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/19bc7271083d423b914194c943e8d217 : SUCCESS in 8m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/509eacc906284e80a39a452d35e33ea3 : SUCCESS in 10m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/df4ec4e9ff544f6e9ded3fd9eecb6dde : SUCCESS in 4m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e40a3ea87df14b25a9010945039566e2 : SUCCESS in 12m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d2b763bb38cb4230ab94c0fcc4be057d : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f9d4f962d8944a30b02143533de5cc5f : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/81241bba9c6143f6bd77b3836fd706f9 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4383cb3ffd7541a1b8a1a60c776fa6c3 : SUCCESS in 23m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/083a941fc55d42a48f4317382d216e5a : SUCCESS in 8m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f3d68008533a4beab3e0f78fac6957e9 : SUCCESS in 17m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9b69e7f8f89e417a8a6f4191a9b3c763 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99c14eb98167443fbe867983e909e4f1 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41629f71eb9c4803a6b2bdd749f739df : SUCCESS in 29m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/09c8aa47250e4d16bf62b98066a19f4b : SUCCESS in 7m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/932e0501e9874581987ccc5d48afdd76 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d0ca4b99d1d43a88d7d6b9ad4ca1c8e : SUCCESS in 6m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1bddbbe887ed4f3c8d2d8a0486ddcb62 : SUCCESS in 16m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/132560a42c444616ae85dcdebbb1831f : SUCCESS in 8m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e3685fdea2444bdb78c7c970f82e652 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b90d4977aeb450197ffd4ef8a11b474 : SUCCESS in 1h 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"31558872d89ca08d02b09f39086898200fa923d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-20 06:34:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15a1a4cceec0c20808cf577d396832164cf285f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-20 06:38:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6545de8d23189a0ad3cf3e675bb76cc7e16ba5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 06:39:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52366ecd67a171b7d3c518ae1a9ec7563fbe7215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 08:25:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83598d79f95a410583f5a4bb872a4c2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fa6269ce94437e92538c9aa46a26ff : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b82d4a7e5354f78afdf443c5e2e9cf3 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a57507a330ee43a6bc5cc5eef34252bd : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b477882d23a4e3db6b0675d97e450a8 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8eeed812b15245f78addb59df5fa3c57 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94f0a91fe6514c53b79b0f9afc295cf1 : SUCCESS in 4m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e86fcdba4e5a497ebb49dfa6ea3ff289 : SUCCESS in 5m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a562932ec2cf40b1838a31362928d20c : SUCCESS in 9m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9e28dd9083f44487836957e3fb96a02b : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/327955d5d47f412a998c44584b4792fa : SUCCESS in 11m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b6568b9ad5504839a200393c9cb1a044 : SUCCESS in 10m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7e95c3b922ae45caab1d4539fc251f58 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/28c10c1477fc42c58516c6df9a900a38 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/27e69e125cc440cf9aaf1e57490b2043 : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ea4ebc10aca84ccebefc53f7b67cbf82 : SUCCESS in 18m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e717dbaa4e354881bd6f0db14b1d4848 : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3b81cf9cd8094ff194f82d93b1015910 : SUCCESS in 7m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3adf1138868a4ccdbe915809e9172313 : SUCCESS in 10m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/893ef23375f24e4383dee370f76e7c5a : SUCCESS in 3m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2041b3f60b9c4a29ae90c3a95a3f5cff : SUCCESS in 9m 53s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2153c6703265468283de209fa37f3f7a : SUCCESS in 22m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac74ff34eca346e4a87031427db972f0 : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/576eeebb20604b0a9f86e0e56454a898 : SUCCESS in 14m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/30ca3abde36547d8b76eb8a356af2cda : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cccd0b0c8dbb4fed9f580e1a159f9387 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5d977b2b8e94775a2c6b7da60e7da74 : SUCCESS in 27m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2b022924c4e64760b3ea50612718ed90 : SUCCESS in 7m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/775d1b2c9dfc438089f3e9fd22b08366 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23b9fc6fce4b433ca22d51b8b9b026ae : SUCCESS in 5m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4a5fdfd88d5a4e4daa32cbef8d9d3c68 : SUCCESS in 15m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7407b37f920b4726b87ab51db7cc47c7 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a117733910b411e80240b2c41870369 : SUCCESS in 1h 09m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db3695fe8cce4370ad275510b0f0ad26 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2aa473d885d1879514731a9542cc2bc4d7ad0be1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 08:25:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08109e65d7a06a08fb2633a70cf401540e85f35b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 08:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f08a2f753db4518a9578dbfb19663f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8a754d67694557b4ee1f4c82f4f9d9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15c554d9e2bc88125e660acc1b00bcfbbfcda0ff","revisions":{"15c554d9e2bc88125e660acc1b00bcfbbfcda0ff":{"kind":"REWORK","_number":1,"created":"2026-03-13 15:54:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/980479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/980479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/980479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/980479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/980479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/980479/1"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-13 15:44:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-13 15:44:16.000000000","tz":0},"subject":"update constraint for horizon to new release 25.7.0","message":"update constraint for horizon to new release 25.7.0\n\nmeta: version: 25.7.0\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Commit: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Change-Id: I52faa5ecf77c5023b5eea0a977c7c62d7a9b9731\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8543c244f325d6149d4701db13e1f6d23b0d5796\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c554d9e2bc88125e660acc1b00bcfbbfcda0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c554d9e2bc88125e660acc1b00bcfbbfcda0ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
