)]}'
{"id":"openstack%2Frequirements~981745","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I0d9db451d3e5664e302fa5d859a821fa5692529c","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-04-09 11:42:39.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-09 08:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"38329":{"account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"last_update":"2026-04-07 07:08:34.000000000","reason":"\u003cGERRIT_ACCOUNT_38329\u003e replied on the change","reason_account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}}},"hashtags":[],"change_id":"I0d9db451d3e5664e302fa5d859a821fa5692529c","subject":"update constraint for keystonemiddleware to new release 10.9.1","status":"MERGED","created":"2026-03-23 11:46:38.000000000","updated":"2026-04-09 11:42:39.000000000","submitted":"2026-04-09 11:42:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981745-new-release","meta_rev_id":"08ec4f88c9109d003101b990bda6951c27a081e0","_number":981745,"virtual_id_number":981745,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 11:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-09 08:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-03 05:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-07 07:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-03 08:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-09 08:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 12:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 09:19:51.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"CC"},{"updated":"2026-03-31 18:24:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-03 05:33:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-03 08:34:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-03 08:34:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-04-03 08:34:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-03 08:48:25.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"REVIEWER"}],"messages":[{"id":"c99c5f28b6b7e68a597dc10834fb5cc6c7aba0e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-23 11:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d22248fed3752d87c57e673af507716929b2fcb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:43:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/931ef04b32234876823381a4593e7923\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6397bf1dab4f31929acb0f2a203e95 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a59b28e3fd1543aa91d618bdf6a9b42c : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2500e73903de464ca242a45234c613d2 : SUCCESS in 2m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7117907df44d47439b6d49e2fa724be2 : SUCCESS in 2m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c95b2da05bf4b12b5fec3d0869144f2 : SUCCESS in 2m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/beae7e48e0474746bdefa2178e818536 : FAILURE in 8m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/874e25f739f9429ba36cda746e11363e : SUCCESS in 4m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/21a20ac4453e464ebac06369155e90a6 : FAILURE in 9m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7828301bb81b417f992d49c3147488e2 : FAILURE in 3m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/504879b2304342f7b61d1ddb6119be32 : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/deaef0880346454eab3811c2411c06c3 : SUCCESS in 3m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e141b8e0837f4e65932a808fa42104b7 : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1138522c0bea4463a1ca7947d24972d4 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/740d53909a49466cb5423700d510bdf1 : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cadc7ef0d69840c9b35be3a873834904 : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/66906d3d1c0d4cb486281f8229dab17b : SUCCESS in 7m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/74a4625191f2409db9f10c22460e0cb6 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bbb45454bc547d7b05ad9c4778db211 : SUCCESS in 4m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a7d2e0652a8d40aebbd293c789f0a9df : SUCCESS in 9m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/18fad9dc0f94417f80f7a19c30c0a0b7 : FAILURE in 5m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/54f1188e053c42beb1767bf0f92b26b1 : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c57d71a6217c4cfc85b88e061538fa72 : SUCCESS in 10m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9e0b97d669a74ed5ac782bb6a35cf069 : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ca4d6bf09acc4ed58f4cba09f7674579 : FAILURE in 3m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c58bda5707a845e4a4af83c9985d1da9 : FAILURE in 3m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3d2403b638ae484e9a431b3b7a0c2e68 : SUCCESS in 21m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/16d8038577294fc4b2547d26981a0f61 : SUCCESS in 8m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4148e0d402444126be0ff7e53f98851c : SUCCESS in 13m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ab0e54661d8d4f2abc823133b4f0d783 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7c8dbde1a954427a50408a07afa10ee : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e89ff63842c42fa97512453d2e9a28d : SUCCESS in 25m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f50a3279612945a19027a87824398350 : SUCCESS in 4m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e6bed889f87a4a40afb3c4d221c9a320 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8612c06c9d0c4fbbb12e5d13d458e372 : FAILURE in 2m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1dd2f8df1e534bf58d8bab5f925a874f : SUCCESS in 7m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abf365c42b9b4e90a4ceb3add1503f2c : SUCCESS in 32m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/049f4c0810c6403b94bcc557d4b4446a : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"545271382621699be185839d9e2c2f47ad65b59d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-31 18:23:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"f38805f5a376a0c462ca7a4085b70728b69da3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 19:37:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1da7eced09c461bbed2b4fd796546d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b616bf99424c9084800884cf48dea7 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b97cd424254fd29f7aab77efc946b4 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0acda11d5009459fb4a7ea09884f9db4 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43ae51ed74eb47eb8a4d27de95a8e080 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de282cf4636441afb1e13defd636f851 : SUCCESS in 2m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3e1ef66c20254dfb9f050e3c9660f722 : FAILURE in 3m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a15ac97d5ba448679e77eb699089c7b8 : SUCCESS in 10m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/709cff5062084d99b414cd780dea28e6 : SUCCESS in 17m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0ae96ddf04e04827b0bafba61ac06071 : SUCCESS in 8m 54s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8fd1b5a447114b9e99ea3c6c57ac33a4 : SUCCESS in 2m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/838b9b9030254c3facb38484795b42d5 : SUCCESS in 6m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fac9e144c5614d41992acc53c58f2a12 : SUCCESS in 2m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/52c8dfbb648a4d4f870c3432646fe7ae : SUCCESS in 7m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bba646e6d6f4426e928823ceb9f6bbd3 : SUCCESS in 4m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9cfa7cad549a4e56a5ee56f3e6aea2f7 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2305b0eee5694af8af82ad6087458190 : SUCCESS in 8m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/da663a34c5e24b098ef5ca293c372345 : SUCCESS in 15m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46dab59e1c7e45e9b79311ae6e5ad906 : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8cd9cc13b5ee42d8be936a9b935071c1 : SUCCESS in 5m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/69311bf17f774ef5baf6ad2aa9987dac : FAILURE in 5m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/38ffc53fa3b34194ba048d3ae00b9d70 : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/876d4fb7017d43399ca267c478a14f1f : SUCCESS in 7m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8c39650fed47467a97ccac8e38ed88d7 : SUCCESS in 5m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eb9b7cd92c9240d795bcc718dd2f169d : FAILURE in 5m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f69f037d751f434ca941e5442a56e2eb : FAILURE in 2m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5233b83e27de4565bcc1d6fd8fcfe0c8 : SUCCESS in 21m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7429227b60b3454bb6016f7f4aa4dba8 : SUCCESS in 7m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3afe35ccbb424eba9d238be8794705cd : SUCCESS in 12m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e0b7dd86914a4635a0b6bc8927f66b0f : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5fd5c07d0fc54f9a9432cfaf72397b07 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acc005496d6849818aea76c0d9c4fd54 : SUCCESS in 21m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e6aa09c6d90d4620b7270848f76a3304 : SUCCESS in 7m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8ed3e77f9ebb485fa44e2ccb6fad0f00 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/450af30142924b1d905099909571a4a2 : FAILURE in 3m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/40a8becb285848959d4885c8e5a2f710 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a643d0449daa40808cb63e70df53416d : SUCCESS in 1h 07m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/183c0df890bc45779411eedabf832327 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d5e890ada8cfea92071d5f75038e9e61b53227f3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-01 06:25:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf61e7de88835c10ac5ec896d39c8d2100a5e72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 08:16:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd1e33bd6f894bfeaa64652bc8bbbf50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1e57c407e6459fb2beb431d524f71b : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4193b4ee4c74a8fa7173acad5b47907 : SUCCESS in 8m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0f669a24a3f4e019501630317cc0969 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d39347a3bc274261a21d47711e718d20 : SUCCESS in 2m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33620ee23bdf406eac4aa379ab089d02 : SUCCESS in 3m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1e158eafc0a48acb91855875a7e62ec : FAILURE in 3m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/41bdbafaa3bb41d7b1e50e811b7a3eef : SUCCESS in 15m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/65a54cc6fce940489569650810ab532c : SUCCESS in 21m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d26f99cb493b4ea893bf383894f6e342 : SUCCESS in 8m 24s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e5479862e5e34f64b0a851a0c3e8b866 : SUCCESS in 2m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5ee918ca4efd488e876ac6f8fb7b0292 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b95d720a059143da86f5fe6af7d4dbcb : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/41ac955f24a84011bd617f9569a2932e : SUCCESS in 7m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c84f1ce627de4d81ac8d32d3e8087eac : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/918f6ee4a6d142c6b55419c37d826627 : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a587162ed88646e2981bfb15bfc607a3 : SUCCESS in 7m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6c0bc2fd20ca44c8821cc160dea521d6 : SUCCESS in 14m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/019804f64bc345cd8dbe805bff7d1d8f : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/62e8f589e5ca4ce79c1c4b78458a04dc : SUCCESS in 8m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8a1a14ef6996423882f7c21126280c38 : FAILURE in 5m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/23f9c411b8bd40edb0d7d399ee2caa96 : SUCCESS in 1m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9e8fb3f000654a4889b829c5cc8ff349 : SUCCESS in 9m 55s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/37be4bb700794722baf89c9efbd6eb8c : SUCCESS in 4m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ecbd38844794415995cf1fb041339bdc : FAILURE in 5m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/26b2e041bfeb4d7fbb47c03c03af247a : FAILURE in 4m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/68ad8fd0386848b09264e20d1379c72a : SUCCESS in 21m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac5d934d770a46579f635f63b0434b1b : SUCCESS in 7m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0a2db52a2f7f4bd7a6fbf98bb92e010a : SUCCESS in 13m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6b3aac3153444dddab679b67b380d599 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68048175d2f448afb0039f56d3a99a0b : SUCCESS in 2m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bb0fc27c0944e15a92ab44e197a0dbd : SUCCESS in 23m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cca08421f9324eccbd934f158a8469e1 : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dec9be752d704ade933ff37ee43a6dd2 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/febc531f5a3a4b71aec21fe0921aa7e6 : FAILURE in 2m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/27ea39aee2a94433987f3cd447c89a90 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a29d4f208b094b61b8be98165704a9ca : SUCCESS in 1h 01m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbce4a591d6646c098bfd7b9ece28c83 : SUCCESS in 1h 49m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5e11ecd33fa343c5a64f2a1c56dfd75198aefb51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-02 15:23:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":3},{"id":"f87ac99ac141ffd53ef037dddb19842cde2ee52d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:14:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec96e216c0c64f0a95346ff0a9745c64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed86a837fa74a0784b11288f87de939 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd64ae5fb7ae4afc816bce8e9c4c9254 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ec1703c91524bac85e52e6c4530e9b2 : SUCCESS in 2m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/421be7593307446d8e43da097a02b341 : SUCCESS in 2m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7462f2209fc455ab1996444d8a610fe : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1d66e9d12d494fa4b77f4a2e3569775d : SUCCESS in 8m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c69b570b56cd46fcba6dcf1b609ce7b0 : SUCCESS in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7efed365503b44faa4589b8152842b9c : SUCCESS in 14m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9a8d1baea9004a44b89a1de65c80dc94 : SUCCESS in 9m 44s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c3e4bf7c143a4c659e4a0b444ed55903 : SUCCESS in 3m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/58b062e9f39c4a1e94fdc9bdc2b3784d : SUCCESS in 3m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f03cf917e7d4423e9cc2850fc4f8cace : SUCCESS in 4m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/eaeb6a0251d64854b2602ae523d46964 : SUCCESS in 12m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4c28cee17a5146f8a23d2a646567cee0 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/529ab84589dd4b46b07fc4f85e03e8f8 : SUCCESS in 3m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e35b28e4ecca467badaff9bd52f04fa5 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/622b673ff2204cd1823790f95d97c699 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45647284dfcd4961aa9e4c5a264111b6 : SUCCESS in 5m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f2e17d6c0da6411499f3e0b13240ead5 : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b17db3197d0f4a6ca643fb63d53b3962 : FAILURE in 3m 14s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7c2d1f2b05bc46b9a13ce3cf8b68711d : SUCCESS in 2m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9b8d6b762f17416f99d94c5d5762bf8f : SUCCESS in 10m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7440642baf914c7e92b87c83ee368785 : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7ae54740c38147b887e34dcb1b1acac6 : FAILURE in 2m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/871282d0740e428b96462444701c9389 : FAILURE in 7m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/35882cca32c041949e309656a4022d01 : SUCCESS in 15m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8a8ef8116c5c453db50d119f33171b46 : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a80d3e74e27e40fcbab635387deaddf7 : SUCCESS in 13m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7129f257c0b643cfa65135c3aa0e4e48 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b31a7204090e4436b71f8d1fb2215219 : SUCCESS in 6m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4019abb961c940689eeb16289a4419c4 : SUCCESS in 24m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a0c2dc2c7aaa432da7bbbfd3534d8d0c : SUCCESS in 6m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4c4fa1b90d454a6a80282881d331df89 : SUCCESS in 3m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96c828ec2505419da9709729f3132831 : FAILURE in 2m 45s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/19b713aa2fa34aed8c28db9336a17805 : SUCCESS in 7m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/420705a68cb14a7ca27f78d76f22fbf6 : SUCCESS in 1h 04m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24ec34949dc74b3caab6b277b479d7b2 : SUCCESS in 1h 44m 45s","accounts_in_message":[],"_revision_number":3},{"id":"ee0f21646319aa9db37c36dde3a567e432183293","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:33:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bce05e6db65631cc0f53eb15f48bda24be57477a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-03 08:34:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d67b32d49f9535b4b8044e1bbf937f1811544c93","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2026-04-07 07:08:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e9b44f4033d3ff5d0e713f32119ad513def3e594","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-07 08:06:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"90f7c7aa626038a40ebd1da2f1c5af08a7e09cda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 08:06:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f289790102ac94a5c609b68f85f8558562b048e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:49:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/922fa7950e2146e5a6860041c07a6e35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ff4591f09048b6aa2d0ad9acfb3622 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/708176581e5341918b4d66ff634c5443 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a1689069e76471285ef5f5856bb38c7 : SUCCESS in 2m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4b51bf985844de5931e1ecaba2d7dec : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4c41da2abac47b2aef5266cac1013df : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b3678433f4d4c38968e2629a6549d40 : SUCCESS in 16m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed21c1834a834139a257b9b4ae333bae : SUCCESS in 16m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/925b8564f9624b6fb37618f1da8d2492 : SUCCESS in 20m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/10ee7de849cc4eda819f770522f421c2 : SUCCESS in 20m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e744010026d2479a932249bb6ec193de : SUCCESS in 13m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1de620c833c94e58932c688855343d1f : SUCCESS in 3m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c7b94527ab3640ea91a194223a8255e3 : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3979edf9c84e4f34801f7eac3cd260f1 : SUCCESS in 4m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4cde3581f2f143f3a2d4cf380b629f4a : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44f95622ce794c6ba353435868b3c3fc : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d22cde57144f4565a1b710513ce1fdd3 : SUCCESS in 5m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/411b469259c44a5c91992e0bf37dce14 : FAILURE in 5m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/59e44ff7d4f44ce6befecab1972a7fab : SUCCESS in 3m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/732c445415ae48bab812a1c8eb19d7b5 : SUCCESS in 10m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/128ddb8623d544dc9d9df9b268fa354b : SUCCESS in 20m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/be6cc66f449e4416a7d1c76ee4af9a59 : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/38d413c7bf984582b1927055e7e91612 : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fc91135fa2d2473d80f1e5776cde404c : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed0408c542004d5bae0148786d9a02c6 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20695455147345c885018471e9b3cf06 : SUCCESS in 27m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01a2af2a320043d1af4ec1ad3cbf3f15 : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3ca6e0fada324a9289554f8b5a77def3 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e730caf630a9421d956841371f6a0de5 : FAILURE in 4m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/284895a805ce4c1a9bb7bdb2f0e50b6e : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7751cbcdde841beaeb02f967d6829f1 : SUCCESS in 1h 05m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb1feaf0a5b0412a921a4837793b257d : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":3},{"id":"04181da0a788728abd66ca2fcab442dbd0ec1c74","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-08 09:23:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"880b2617bfa33db4cc02521f647149980d9a9cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:31:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92830d66ea574c048336de7f05815db5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d0867c06d448ad928e8f9c3948d66b : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a23c23be12740efb5b32687626dc3da : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65ed6ebf79db41cdaab592ddca0fe71e : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e13431af193e47149d5542a53e668895 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb7898a41eba4431a1056222c04ae354 : SUCCESS in 3m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b7cc89fa59c4f0293dbaeb1d64e6a3a : SUCCESS in 6m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/385717b2b13f40368eb883489eb37618 : SUCCESS in 10m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6e0fcd8daa6643ef9b00f195c119705e : SUCCESS in 17m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/54b8d54335c04b6293cd69e1547b93b6 : SUCCESS in 26m 25s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0c8688706c7e40cebf77db8a707967df : SUCCESS in 4m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c9010a3b595342d8b4471d57bb727c36 : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4ec10b448f0f4ef298f517d22e3d9dbd : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7232d6c8ec924d7898739c2e456bedc7 : SUCCESS in 7m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fdcfdf662fe8415097418d476c49e130 : SUCCESS in 6m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e0eb3112d0274d188d189c1c8eae92eb : SUCCESS in 3m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2ab127f3bfb0434ebfd44c99d64e3553 : SUCCESS in 8m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8a9e2aedb11246e1a3b5bfb895115200 : SUCCESS in 15m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96fcc522d33f4c2ebb39f0fe05519a48 : SUCCESS in 7m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b9c3870e385348a98af4bf82f4ccebf1 : SUCCESS in 9m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a3c51c726c85453eb8a19e972c9d4bc4 : FAILURE in 5m 33s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9aa725f85b754e1dadd722b3020f8df4 : SUCCESS in 2m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9b42ba0e3fd74afd97875aad6bd7bd3d : SUCCESS in 10m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4c011eea733742469487747684b0a9a1 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/abc4cf2b1e884d51a23887f666f98c4b : FAILURE in 4m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b0e6fbcc48d043f99694b92f22390ed0 : FAILURE in 4m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c11dc77d9d264ee7a9d76871021c4889 : SUCCESS in 21m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c2e5e05986e24f089cb05d58ea29b890 : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/089508874cf94805930db628455a162b : SUCCESS in 14m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4929c08d996a48378fde27169619afa1 : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a240bacd33345ef9f0309fdd94aecba : SUCCESS in 2m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a83e59d3c4244a9a0212cbcf0c6a48b : SUCCESS in 24m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/be03ea25b90a4259ba17e3768ac6bc4d : SUCCESS in 7m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d6b54ce3511442119700170de8a5bbab : SUCCESS in 3m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d77170518b26416c890f0c51b1764eae : FAILURE in 3m 05s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2868a368b31d49e781b5f56d2c60a5eb : SUCCESS in 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b75c1d0978b4e049b5a5894d907f38b : SUCCESS in 1h 01m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c1d205aaf1646e28acda463d14c6829 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":3},{"id":"3c524d5632530270320477ca9bbe78058a844dcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:32:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c5df768e934144409e9481886a49625f43d1742a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:25:14.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8f5c64df8c2485ca4dbebda16923c5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671379495f0347f591f7a1fba872356a : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f31af745d9264d9b8a76c923a1fe5cb7 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/395c2918a74a42788381cb343c8a7acb : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/828816acb7be4689ab8e8a51cc083bb6 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/506768bd977f435291a245f4c37f0ae4 : SUCCESS in 4m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2d376e2afeba4409826e70e088d4b2fa : SUCCESS in 17m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/01920591410b498db7b775f69bf11916 : SUCCESS in 9m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/29b253077777487e8f1827b47338db62 : SUCCESS in 17m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec6264629cae49b19f77459981642a51 : SUCCESS in 21m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/69fb19d2c93c4f3ea034edeb3d6dd069 : SUCCESS in 12m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f2e69697bfc84e7a8fa76d1bc8308803 : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5ad2a011ebed462b94aff8c98c0912b3 : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/64aeba56d69e436f9766e2bf83636b56 : SUCCESS in 7m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2e0d842b49024b66b41178b189873e14 : SUCCESS in 16m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/561c5e5ae54f4f41b01ac8b411e79e45 : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7230b244f40647babd752d045e21c4af : SUCCESS in 8m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a37a25aa859545439f4b1541cd6730f9 : FAILURE in 5m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dd57617488a34f7fbfb2b2dd64862dab : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8b5c0037d6504f5f965a9edcec61c527 : SUCCESS in 9m 01s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a642b8d897144c1ba3531c20bcda9077 : SUCCESS in 24m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ee09a4c815b41679efb01e2a74cd1ab : SUCCESS in 4m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/61a683bfb0ed4de2a84bbe7915c2b67b : SUCCESS in 13m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bc822aa984a848af808c5d961c529f1d : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83c3676382724cfeb7547be99075fd74 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a36d0a1c5fa1471aa6bf459f5427a0cd : SUCCESS in 27m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/09f65517de604c268505f29bc99b1929 : SUCCESS in 7m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9cb4b05c3b454a55b1fe57e62f6889f3 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10b6cb10cea84b33ab5de7b0d7dd5517 : FAILURE in 4m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3450ee3838304ff0b3f3f5f13f277b5a : SUCCESS in 19m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eae062b8d57e433487d0ddf4cc4ebd35 : SUCCESS in 35m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb5dc6d6763a42eeb7a2e8a7f633d238 : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":3},{"id":"a9339931db97623d3c6b3955ae184b0d4e994367","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-08 12:48:06.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95750d2e88f16df8858293ec3661a856aea6b6f6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 08:15:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0104b14667c77552a64f253d02f8c661a235609e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-09 08:27:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f13c99bfa1340566eb57a08168c3c2794808e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 09:49:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1055bb113fde41c0a2ebf5daea7d6749\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d9a8271eb34085971d39c1a72b6de4 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9937f0b23ab544ca84300257e1ab9442 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d60c506a474a4b82638e06535785c9 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/132104e533b54084910b08bb306973a6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/990d22cc88ad4fff89337904a6224d70 : SUCCESS in 4m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6d6250cd1c694d1697f286db9bf26bcb : SUCCESS in 13m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e3cdd7b63ffa4121abba876217227bcb : SUCCESS in 4m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e9eeafc96a184352ae7de3f39ac93bd6 : SUCCESS in 12m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d24fe6a76be04e20b7370e51c3c044c9 : SUCCESS in 21m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/96a5e8224f9b46589249eb27aa040506 : SUCCESS in 3m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/931c28c0ef5643b7b9288fac54ba2b11 : SUCCESS in 6m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ca19363879a54aa4a51529fc1c2ff001 : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5e419c76e9f94921a9d36c26f75d2ea1 : SUCCESS in 10m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/628fec2ea9e549bc87173227f03ed5e8 : SUCCESS in 5m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5d991c928fd74aa4b1a046e10914dd88 : SUCCESS in 3m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ad6bceb6951344d7885bc3f25890fbb0 : SUCCESS in 8m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/80863ab2986741e389407c33c092b818 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d7d494ed8814744890b68ef20920485 : SUCCESS in 9m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a91138accc5349819f38a9980da204dd : SUCCESS in 9m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/30be71bfeb3b4b769d083484926185e2 : SUCCESS in 12m 50s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cdb7c4c145144360a8193c9fe29194bb : SUCCESS in 2m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/871459d1ce7d49d892e238e3fde2c72a : SUCCESS in 10m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b5e003218b254e559460c8ae8d5908c4 : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c04d535c7cf840c09c31419d7b659386 : FAILURE in 3m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/554c86a7ce7f4cc8ada43f6728944e80 : FAILURE in 6m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d5f6d0287f35408ab73afe1fab00d319 : SUCCESS in 17m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/356a328d9f024881baaf935c32548a44 : SUCCESS in 7m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c69c3e1d9fe24eb98c8270d629a9df89 : SUCCESS in 11m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d170dc7050f348acabbfe6cd7232a4b4 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/17fe94560c35482f9376a5dc86c34486 : SUCCESS in 5m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a34c3858daec450bae24b49520be2ac9 : SUCCESS in 25m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a7f729bce7a8460bab0d20c6066fb403 : SUCCESS in 9m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/150b156bf10446d6a62f2ad07d41387a : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7991ec9807c44a8b85ebe6fa51c7db7b : FAILURE in 5m 08s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b6e95232f7974d6d92bfcb6fd697888f : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a07ed65edf84326a356e120bd590dbc : SUCCESS in 56m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f4d74940a534e46b0a4350b347572c9 : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b7e9f36f79e9257042d7e176da8ba74fc6685165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 09:50:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"61fb693fa03f800b2ea1e53ffbb74fef00c95ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:42:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ca3a64911bf4bbf90060525a9ab10bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a66eb879bb174787ba0c9114ff3ed401 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24d3fba402934bc492ca3cfee03d2037 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4b42e1d425e40a6a02f186039ef42c9 : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffad7bb600034b8ebae74fa9a755522b : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7db1d687ae14ad98842a53aa79758d1 : SUCCESS in 2m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/78d08a901ae7410388582c66a7549f82 : SUCCESS in 8m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/87948fae409c4de8acf94ec40ab1e2b1 : SUCCESS in 5m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e2154352d8e44dd4ac8ec6c6217b6845 : SUCCESS in 8m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5b5cca65bec540a9bed6585fdd62cf72 : SUCCESS in 9m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/69a0d9d1ea2f4cafae6543a82d956fbf : SUCCESS in 9m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aeab7204c8684bd5b79a87fdd5e63f02 : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c688979e414841329db3640ffaf5bff5 : SUCCESS in 4m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a3ea567aea6a43219b4d76c19cfd7109 : SUCCESS in 5m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e4bd37550363452e97c0e6e48f3d48f9 : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9df213f2a3b04f8da92284c557095712 : SUCCESS in 4m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/70e0f910989e472a9f1c3f273dd4cdd4 : SUCCESS in 5m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4e265f1e942e4441b7a1a23628e64df9 : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7bd259f1e1bd424594a156c3bd0cb67f : SUCCESS in 6m 32s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8e4c22a25e814b5a82be0a2b398e8776 : SUCCESS in 14m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e4abe71411e4a96b23511607af1066f : SUCCESS in 4m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/865d4ae2176040e99ad7ece01bce5106 : SUCCESS in 13m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5ffb9349402e4eb598b40278780b47ff : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21c3fe4de6944d00a4aa992488cdbfa4 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57b32c897f134be39d23e55f2ec28726 : SUCCESS in 24m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cb5b823baf204a80add1b67c4ce485a8 : SUCCESS in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/29922c14a46447feb390bc17881f5378 : SUCCESS in 4m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/049e65a52b3e40cdb30b44ab236fd9a1 : SUCCESS in 8m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cbc0f7bed174478ab3b228868a19a8c : SUCCESS in 37m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/439aec74c565481e974d9e9807d01274 : SUCCESS in 1h 47m 36s","accounts_in_message":[],"_revision_number":3},{"id":"08ec4f88c9109d003101b990bda6951c27a081e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:42:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"378bfd3a46066752e3cb7229f3f543457923590e","revisions":{"08c443cb0876988eb3b182dd72d62cc49c4af97a":{"kind":"REWORK","_number":1,"created":"2026-03-23 11:46:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/981745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/981745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/981745/1"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-23 11:36:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-23 11:36:37.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.9.1","message":"update constraint for keystonemiddleware to new release 10.9.1\n\nmeta: version: 10.9.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: I1f8b08e23698eaede7b3502b76129ca0d79d19e0\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pavlo Kostianov \u003cpkostian@redhat.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I0d9db451d3e5664e302fa5d859a821fa5692529c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08c443cb0876988eb3b182dd72d62cc49c4af97a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08c443cb0876988eb3b182dd72d62cc49c4af97a"}]},"branch":"refs/heads/stable/2025.1"},"bb6badb2d8fab289f4c77666d265f69494a1c946":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-31 18:23:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/981745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/981745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/981745/2"}}},"commit":{"parents":[{"commit":"277424d1e7d5a9b0b7f2bdb126d4797232686322","subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/277424d1e7d5a9b0b7f2bdb126d4797232686322"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-23 11:36:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-31 18:23:55.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.9.1","message":"update constraint for keystonemiddleware to new release 10.9.1\n\nmeta: version: 10.9.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: I1f8b08e23698eaede7b3502b76129ca0d79d19e0\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pavlo Kostianov \u003cpkostian@redhat.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I0d9db451d3e5664e302fa5d859a821fa5692529c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb6badb2d8fab289f4c77666d265f69494a1c946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb6badb2d8fab289f4c77666d265f69494a1c946"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"378bfd3a46066752e3cb7229f3f543457923590e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-02 15:23:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/981745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/981745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/981745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/981745/3"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-23 11:36:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-02 15:23:32.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.9.1","message":"update constraint for keystonemiddleware to new release 10.9.1\n\nmeta: version: 10.9.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: I1f8b08e23698eaede7b3502b76129ca0d79d19e0\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pavlo Kostianov \u003cpkostian@redhat.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I0d9db451d3e5664e302fa5d859a821fa5692529c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378bfd3a46066752e3cb7229f3f543457923590e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378bfd3a46066752e3cb7229f3f543457923590e"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
