)]}'
{"id":"openstack%2Frequirements~981872","triplet_id":"openstack%2Frequirements~master~I6f5c61f64f45b4836b201ad2d8e9f629b461f19b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6f5c61f64f45b4836b201ad2d8e9f629b461f19b","subject":"update constraint for os-api-ref to new release 3.2.0","status":"MERGED","created":"2026-03-24 10:42:38.000000000","updated":"2026-04-01 21:05:40.000000000","submitted":"2026-04-01 21:03:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981872-new-release","meta_rev_id":"f0fa42479a4b5bf4e4ccec81ea0f3add59649bc7","_number":981872,"virtual_id_number":981872,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 21:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-01 18:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-01 18:37:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 11:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-01 18:37:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"824bd47931d32be6b9576f435df4ec9b513867a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-24 10:42:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e3e8c4e194a79723a713e9d9414af5aa4576acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:50:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72ae7279e5f4b0ba35295355c792175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88690ab44fe749f38b8e5c52c3291c1e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686b0d5bddfb4e41b34b82fe5ecb5974 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4dbbbf8e3214d1e8f9d13b5939f4fa6 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bf90737dd6941bdb3e7c63270fe48a7 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f369676c8bc4ec6b7403edb5de61677 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccf402a27d764a7e969159298058069b : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1942964316094d669df43517c5334742 : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/407d89dea9a14fa994466f07ae9ee25b : SUCCESS in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/99c71c2adbc1430b8b1f7043e1988859 : SUCCESS in 8m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2f0e521efdc44b5392401aaf7ce514ef : SUCCESS in 6m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/81a9b5ad25fb467da22b5304920ededd : SUCCESS in 7m 39s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/804ef000329b4d61bd17b3be3e13e64c : SUCCESS in 5m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/de5149754d9b4a6a9392442d9bb83f3b : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/60e057727c674c809c40d1e6583e67bd : SUCCESS in 4m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/485d735dc9ee4739a1b0eb8888977aac : SUCCESS in 11m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/82111e4570e24f24a120abe95ca93989 : SUCCESS in 4m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cab37c6e9c99408f8a5d53ca35f21680 : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b1ed1c65cf984d80b0d4348e6441a6ce : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f3d933a118ce4098b43c9422ff51c4f4 : SUCCESS in 15m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d8d2f58246c47ed865eeb6d71f0ab29 : SUCCESS in 9m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2b6675359e9042b1871f9d1a01a2d870 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/38dee7508f99472dbb5ead13674dc491 : SUCCESS in 10m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8cd6d790a3764d77aae015b47b53c807 : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bcfb96cedaf24157b95acec6a3db4a83 : SUCCESS in 9m 52s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a7ba7e74e1f044dca64ac9067d29a9c0 : SUCCESS in 5m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4e7d8e4016d046d18e9ccec4288a199c : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/85786b518b4049659823e7ec4115a58c : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8219132a3d7d493fafad79abd7f75dd9 : SUCCESS in 23m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d281cb5aa1b24f149f1ea49c2b985cb3 : SUCCESS in 7m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e6aad4996deb48edb9cd43de714dd878 : SUCCESS in 15m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/27395f91f87e4a7691810a36218c3438 : SUCCESS in 2m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ca23c9ceefc4442a18b5954ebc7c9c0 : SUCCESS in 2m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2491944619d749b49a22b7c44cd57797 : SUCCESS in 24m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/117b2b8f612c42b6b499da9ff4d3a440 : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/01880e2fa97c4e5f85d4f58d9db68fa9 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/748350dad7aa490dbbfeec11b6ea6a5a : SUCCESS in 3m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/55540222fe404842a2ee7f8519ac3d66 : SUCCESS in 11m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e5c1b4f7ba884e659a62a2dbf6f8f260 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91d224f146074a0798df7952aad84d53 : SUCCESS in 50m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21b1f243fcd1440b916eb3226775a2ca : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a96a85bc82f726b5d665d3d156aec5fc74f4897b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 18:37:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e2589c5f0b5aea87358da313816caf1ca999f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:38:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d31ee66eed496d6a6814e431f57d028672949ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:03:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a5c87c02ea24c018600a923a3973dd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3914a4d8f36409eac16d759241bafcc : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bfdd40ba6f4534898a1eb2e28c7604 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ffc1d9601db47d39a3b4251b4368e47 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d73b78d2f6194c56b5b2d27858f958d3 : SUCCESS in 2m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/770993658d95480a869412637b6992ae : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dce76f5e2574aabb807f27adbe9cc14 : SUCCESS in 4m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a098f9c1b9f4711830ba61323a53f24 : SUCCESS in 6m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66e2297b00d34849958d6a38bc32f9bd : SUCCESS in 6m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2b909d1a06ca448a87320fd3ab328091 : SUCCESS in 13m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/11728d97bce04f198bef7743699b187f : SUCCESS in 10m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/467c1b7cc9bd4f1e881cadada72f4195 : SUCCESS in 11m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5e4513dab0134416ae718362c3df3078 : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9eb37725112b408cae5c7f266b5da2f8 : SUCCESS in 7m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9d944e9fb1e9421ab0845f92b1d4244b : SUCCESS in 9m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/160284e9713140c4b378c0744c2c18af : SUCCESS in 17m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee258422a18f43a38a3c614af5957ddb : SUCCESS in 11m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa88f43918a7447e942efd423ba7a62d : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a8fb7cd4aab140169c88699d47be1686 : SUCCESS in 12m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a24ee9f9197a45588b713f725eab2fa0 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac0c8266372e4bdbaa016b6c7fa2b5fa : SUCCESS in 6m 50s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5ec484fdbee2488baabd1cadaa10ea4b : SUCCESS in 19m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6f7bd192eb1e449181e146da5c8301f0 : SUCCESS in 8m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b28e75c32f7d45a2a04936f4010f5fcc : SUCCESS in 15m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/18ee0122258b4491b33168f5a67e337b : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5da7f3c43d348c3a8c5790f20f761fe : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d3f0432c04a40cfa6d924af3e88abdb : SUCCESS in 26m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/afe05bc676d94d96950fef772b9986f7 : SUCCESS in 10m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/654e508b2a1d40a09f648afd734c3dd4 : SUCCESS in 13m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffe7bff5cff94b69a2b5db030d43d903 : SUCCESS in 7m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5c5821fa088c440aa4e3da142a9daab6 : SUCCESS in 13m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5a66d84fc11f41b18309bba270228e16 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84ee650370f84e85a71b9b74b8ca72b2 : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/850fd49b39d644878d3a98b5e22625ad : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2b6f1dce94944dfb8a2ddfe52d8f631a3d6f2148","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:03:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f0fa42479a4b5bf4e4ccec81ea0f3add59649bc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:05:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f587036110443ba0a2471b6c4cb764\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45729be50f0d461289eb915c2e8ac17c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d76fa17fcd84564677d670832379154e34977c8","revisions":{"4d76fa17fcd84564677d670832379154e34977c8":{"kind":"REWORK","_number":1,"created":"2026-03-24 10:42:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/981872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/981872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/981872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/981872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/981872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/981872/1"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-24 10:32:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-24 10:32:36.000000000","tz":0},"subject":"update constraint for os-api-ref to new release 3.2.0","message":"update constraint for os-api-ref to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I551dca13fe85128f4704e71a5ed5a70f9501a99e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6f5c61f64f45b4836b201ad2d8e9f629b461f19b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d76fa17fcd84564677d670832379154e34977c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d76fa17fcd84564677d670832379154e34977c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
