)]}'
{"id":"openstack%2Frequirements~981876","triplet_id":"openstack%2Frequirements~master~Ie71c9e1a9369a4c86e71ab20dd71749fab68c2e9","project":"openstack/requirements","branch":"master","topic":"uc-bump","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-03-27 10:38:31.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-27 10:38:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ie71c9e1a9369a4c86e71ab20dd71749fab68c2e9","subject":"Bump testtools","status":"ABANDONED","created":"2026-03-24 12:16:27.000000000","updated":"2026-03-27 10:38:31.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b59a102e12bce215b22f3c2d56b81216bef6112","_number":981876,"virtual_id_number":981876,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-26 21:52:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-03-26 19:59:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 13:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 19:57:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-03-26 19:59:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7e926940b77b3f1481060e83e504bf8aea73d9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-24 12:16:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27af7442b50b017b5a7ef750e53eb54eff89d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 13:28:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e70094f3bba4ac29d928f2700f9fd8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4a6cb9ba3841ff9c4d38c615e752dd : POST_FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6151174ac843d2a7d97e24123d9066 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09e61ef6b1274a9a86f7a4ba8fcf75a6 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5012c12e5d6f4e7e840fd52de9266cc6 : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/30e490190e1544cdb33d8756798aa754 : POST_FAILURE in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cb5c58ad2444430bc2aaa55ccc79661 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe8ab693d1f94c5ea87c47eb65dbf3f0 : SUCCESS in 4m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a6c7edb5fd3d4fc89150149fe90283db : SUCCESS in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3fb03cc7a17a4f11baee78735e56d122 : SUCCESS in 6m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3747706172ee4ab2b43c484dec4c61cf : SUCCESS in 3m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/94e2c0f262664b4bb507a03fb57dcfe4 : POST_FAILURE in 5m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/eb198bf669c84d6db50beb404634dca7 : POST_FAILURE in 4m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e5b2dcd39e024607a1538cc14b4b206d : SUCCESS in 4m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/175ae37de55b4acb9d9a860a09a1f57b : SUCCESS in 3m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4789d8b4a0ce4e1ba51bf5564e04a7b9 : SUCCESS in 12m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8a4e4f26eab8432d8d1cd373d861d64c : POST_FAILURE in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d4f4e6217204423b842967064ed488a3 : POST_FAILURE in 2m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bef8960d17df447abd4a9f0ec002b5b2 : POST_FAILURE in 6m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/959a9cc8c35d4c2b9602b12f593f1a95 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b7ac892bd98422bbda8cce8d8083915 : POST_FAILURE in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6b8b1faf2edd4f59beb3a340fdc2c95d : POST_FAILURE in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/738f8437d8c54abe895c326001e6ebbf : POST_FAILURE in 13m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d496f02e258c44a3b626e4fd05312475 : SUCCESS in 3m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/da8cdd8ca26441e194fac36e0e1d0ce7 : POST_FAILURE in 5m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/aafefc14b7934c8d8254688415b77a95 : POST_FAILURE in 2m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/aa8bd95c248f4315b8d251f95570a8b7 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/01195862cd974bd58f9e3feb6fd4932c : POST_FAILURE in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/73a771264c45446ebe16719f9a6fb97a : POST_FAILURE in 21m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efc67c34ce6c45379f7f0fe60315a142 : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d7cb14e1c209420fad4c4bee6a67d9f7 : POST_FAILURE in 12m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e2cce34edb9a45368bb111a44580288a : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/477f3856a94d4209959d618fabd289b2 : POST_FAILURE in 1m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d9f58cf78d24b6abc717878ab892b16 : SUCCESS in 24m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/441a884eba114598a7612c0bc6461745 : SUCCESS in 9m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/21905d6631e8496ea5bfc760e1c61f2f : POST_FAILURE in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69ed10f2a80847a895b9526cdb2236f0 : SUCCESS in 3m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/198ab7c02d1d4fa1bf0729bdb5103566 : SUCCESS in 19m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a30ab49760364ab7a4003071c6296663 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67264b25e9074f88bf82a3f0a4315b6d : SUCCESS in 1h 03m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56da83d9921b4c7b8f929b20c764aea8 : POST_FAILURE in 54m 21s","accounts_in_message":[],"_revision_number":1},{"id":"31a5ef158823fcf94ab298e1afaf344b75ae68d6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 19:57:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc21db15a30f34926c1ede1365c27d77d2d09ee2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 19:59:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"530a4fdd88d42b7aa60a151a3bc14b75e0cd5592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 21:52:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2a6013a349f4d6795435f9755142fe3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd6f7792a0242f0b666d0e65cf9fb85 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340da01a07ca44ac88751e47fc7f0567 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/552a6ae709794f1baf8e33a7716732e9 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/381913f8ac4e47e4a11b06d519765cbc : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a8d70043498e4b809125f3142eb7a840 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff0a4df57d644be589d2dde941e7f689 : SUCCESS in 7m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d09a849c64da4709b5cd2774f4803e26 : SUCCESS in 2m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a8c7c12ec3b047a7be1f9b04c8b8e77d : SUCCESS in 5m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66a04535ce0b47f09fc18c22f8488598 : SUCCESS in 3m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2f003a0a93bc4084889e6fb8a8c91727 : SUCCESS in 4m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d6912eb509914814a9eb7aa2db748f91 : SUCCESS in 8m 39s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/96591ea92df949129cc83ec6794c8219 : SUCCESS in 3m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ba16d0c3ddd14870a431c2053663c3ff : SUCCESS in 4m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6a9c2c938d444dbb9178c35bdfdd4649 : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0488bc052c804e2ea3036be4da294abf : SUCCESS in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/68c65684712f4a7aa0a88db2e962848c : SUCCESS in 3m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2a7e7fa9cbfd4b0982906c0e41f84c2c : SUCCESS in 4m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0803def06a6d4760a08f82bcb5100020 : SUCCESS in 6m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/558b5d35e11b4ec9bf794cc734b077fd : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23b9d980d0b14ee7a3eae0eb10a377e9 : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/22dd24ca556d4823b67ed026afd58679 : SUCCESS in 5m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7686028e297b44348ceba597e589d6a1 : SUCCESS in 10m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2ce24e226aa24c889efed43075c06a7a : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c987402c29f5481191e33b37ebb4434f : SUCCESS in 9m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2995efe8390346649adb72c3912dc284 : SUCCESS in 4m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/02680c3b5a5b4555bd6965510db09f19 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ded112fbaebd462189a06030c9ba3135 : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7101d3da6b034805bdc9348ab75e80f8 : SUCCESS in 28m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a552098237b843fa9bcf3f4077c0ee95 : SUCCESS in 9m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e71fe41b7dc44f891fc23211c37e752 : SUCCESS in 15m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a4baaf2d6fdc4ed0b0bda57a45a1ec3b : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ba824642dc64683a57b22e44a3a34d5 : SUCCESS in 2m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ec9c62692cb42dab6f22049fac9a954 : SUCCESS in 23m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/66f4760d8c6648ef943530bfcd55956c : SUCCESS in 6m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f71ce1cd80f64fab88097152fb70e103 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd6c87c8abba44018209a0b7ca14f59e : SUCCESS in 6m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d8a8d32f8ad34b54b8050e99814310ac : SUCCESS in 7m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/13b7d1c1f2fb4ddeb717bac178e85767 : SUCCESS in 7m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f6ed60ad914fc189139c026a1fd9c8 : SUCCESS in 1h 04m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/601f1aca44bd442386aaf07784f8b379 : SUCCESS in 1h 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2cc85329b6582a6a8b3f7e97146f66ef812bbfb1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-27 10:38:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b59a102e12bce215b22f3c2d56b81216bef6112","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-27 10:38:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"570251020ea0c29b268ccad96d809c719ea7c8c4","revisions":{"570251020ea0c29b268ccad96d809c719ea7c8c4":{"kind":"REWORK","_number":1,"created":"2026-03-24 12:16:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/981876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/981876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/981876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/981876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/981876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/981876/1"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-24 12:16:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-24 12:16:17.000000000","tz":60},"subject":"Bump testtools","message":"Bump testtools\n\nPull in some type fixes.\n\nChange-Id: Ie71c9e1a9369a4c86e71ab20dd71749fab68c2e9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/570251020ea0c29b268ccad96d809c719ea7c8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/570251020ea0c29b268ccad96d809c719ea7c8c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
