)]}'
{"id":"openstack%2Frequirements~982217","triplet_id":"openstack%2Frequirements~master~I5cd04d37068daaccbd91401489918666cb1d61c1","project":"openstack/requirements","branch":"master","topic":"setuptools-build-constraints","hashtags":[],"change_id":"I5cd04d37068daaccbd91401489918666cb1d61c1","subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","status":"ABANDONED","created":"2026-03-26 09:43:36.000000000","updated":"2026-03-26 12:19:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"337aa6221ec196f6ffec9fc25a881dd8fc79e7f0","_number":982217,"virtual_id_number":982217,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-26 12:00:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":-1,"date":"2026-03-26 09:45:27.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 12:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f921274df00712f4ca5266c4d42c024cf654079d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 09:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ec122e6249c87a1fccc851e2adcc0b3a76e6bf5","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 09:45:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"13b8697ccf87db2afca8538aca264a07433820e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:00:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb97a7eea8f84a3d99c26f0aae7ce89c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3e6274c248470c98d09ba2dbc3364b : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fe90cd4da741789bc92f8d32557d29 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a53a844db1945c0819c63a49c46d9be : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05e0e736747947bb8b72cc8f95a54574 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f56a020d4704aea9c2b543cc3846791 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7d29ba763994709bad535597c61d1a4 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/631f78e450c945e9976909eae8978ad3 : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/448d3393bb394053b6bfc9fe55e3f997 : SUCCESS in 1h 04m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c11cb3b6da50416f9253e8729aeb9eaf : SUCCESS in 2h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"337aa6221ec196f6ffec9fc25a881dd8fc79e7f0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 12:19:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e31a992cb8c49fe4a54188def0612a439b405e21","revisions":{"e31a992cb8c49fe4a54188def0612a439b405e21":{"kind":"REWORK","_number":1,"created":"2026-03-26 09:43:36.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/17/982217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/982217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/982217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/982217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/982217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/982217/1"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:43:28.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. This is an attempt to make it\nwork for those branches.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e31a992cb8c49fe4a54188def0612a439b405e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e31a992cb8c49fe4a54188def0612a439b405e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
