)]}'
{"id":"openstack%2Frequirements~982232","triplet_id":"openstack%2Frequirements~stable%2F2025.2~I5cd04d37068daaccbd91401489918666cb1d61c1","project":"openstack/requirements","branch":"stable/2025.2","topic":"setuptools-build-constraints-2025.2","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-03-30 08:35:41.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-03-27 12:20:31.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-03-30 08:35:41.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-03-30 08:35:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cd04d37068daaccbd91401489918666cb1d61c1","subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","status":"MERGED","created":"2026-03-26 12:09:04.000000000","updated":"2026-03-30 08:35:41.000000000","submitted":"2026-03-30 08:35:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982232-setuptools-build-constraints-2025.2","cherry_pick_of_change":982217,"cherry_pick_of_patch_set":1,"meta_rev_id":"0e621c776b7440811f7da4b57a79f92c4181193a","_number":982232,"virtual_id_number":982232,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 08:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2026-03-27 12:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-27 12:26:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2026-03-30 06:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 13:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 15:56:03.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-03-26 17:07:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-27 12:26:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-27 12:26:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a42112bbcf81956b1a360f697d74ea4eed507b5f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 12:09:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* tox.ini","accounts_in_message":[],"_revision_number":1},{"id":"ba3e31b1646cc782b23648bae10df4fc28e197a8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 12:15:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c173377cb8eccff3259d8aafce91d5a6734cb41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 12:16:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7ffdf448b48eaf642d58e7112e9ca2cda3bab8e4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 12:18:01.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"02bd9047a43389e403a7c17ea86159074c9bcefb","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 13:34:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2186079bf6c22cd260ae965e81c389ba5616fe96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 13:42:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc373addf4b44009f5fdca254a1175e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0592da689e43ca8632033064deb8e7 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/334d7ad3b5144ec9be027a77b7f9ef99 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3f8128058d94fc88667e5ad86916087 : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e22dbe0ba0ef4f5bbead0c1293209e28 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3581f007956e42b38cd69703c41fa6bb : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb29fe756b164b27bef75c62fba73d3d : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89a1230e23154a4caf80bfa5197ec477 : SUCCESS in 50m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2397187b9ca4277a350c4264bf7d3e1 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":3},{"id":"9fb73d92c8783ca8842ee69832271472ef841179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 14:31:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a9c7745788dc989ff6fcea447395cd0413b2eba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:39:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217d8732c05c468093c3b31ce947856a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8269eaed5a4bfab49f4e6d2859675e : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fc56c3e71ea404e831da28aadbda97f : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84d8d70c65b3410d87c6bfc3073c4f4f : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b250181ba47c413290f03343b5d76fde : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58b0ef6478354d4985935c66675b4b1e : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5a446bd68ca4e5699c9ab6c57000926 : SUCCESS in 3m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddfb646b09a247ca9fee402bfe5cc0eb : SUCCESS in 1h 02m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49ec60bb70ee4855a8cb0feed8c2f191 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7d91ce53be1051b6ead4e508ada1d46b348ec056","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-26 15:54:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c19eeaa051a50004edd87005142dba9f06aaac76","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-03-26 15:56:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"31fcb993236dc1fecca31aeabbdbc2dde7429da9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 17:07:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"32d44d39e13ea99852d1f9f1844b23625330a9f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-27 07:46:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"872e09ab1a76fa8226ff721ac89cf63054757c41","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-27 07:47:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a0b64958abdcf428f54b381f8aab472f2513bf59","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-27 07:47:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e2ae95eca0202e13488f1cf27392abfbbc360750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-27 07:48:07.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"c4d5b4e0e8c51b73d72862e43faac2e30c5462d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:09:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d851c83ad49457a913554f8f30fd38f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2e603c1f5b484ba55d64b1cde4f57a : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a4479b9068b48849a50f89a858c99e5 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d204662711b494184fcec4a35d62fe9 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2305e0813558402e855893ff6f2edd4b : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f930d533766c4e078de290cda3b49389 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/387a3593eacb42f6aa353de972882fb6 : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4e177f311dd461194912140ea8fa9d8 : SUCCESS in 43m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5ec2db927c74ef6aac563663bd92a50 : SUCCESS in 2h 15m 09s","accounts_in_message":[],"_revision_number":6},{"id":"dea27002f744c6efdea959c8b6edc5ce80d88656","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-27 12:20:31.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3973f987a97a461a106c97d0bbc54182fed030cd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-27 12:26:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"5a0efd13388570c2f199fa301ce486802e867bc0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-30 06:58:53.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c3d112c84d4f7be2e00021d0e2e7c6d43001c06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 06:59:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0827b0363a3189298a08cd339daacd79f5e5b51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 08:35:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c13725cf0f4c42a92558735ce2cea3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd18122d7d947a5a1b5c033645d313b : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/947dae2887824ceca89e54c75571dbfe : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94b3b199aa9d48dcab951c1edd08d2b3 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/003942d3a1584e43a3fa87c1b367ee60 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a81d2cd5ded4a248f4b3b1316fb7c78 : SUCCESS in 2m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33cc89f73f07456da411e51e5810ecb0 : SUCCESS in 1h 02m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd9dbc8d3f5543aeb584b460ab264594 : SUCCESS in 1h 35m 08s","accounts_in_message":[],"_revision_number":6},{"id":"0e621c776b7440811f7da4b57a79f92c4181193a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 08:35:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"96854c96d2c294e59727b482ec259db78e25c2ff","revisions":{"5ff3aad16851afd8e8eb894bd50c83121a142943":{"kind":"REWORK","_number":1,"created":"2026-03-26 12:09:04.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/1"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 12:09:04.000000000","tz":0},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. This is an attempt to make it\nwork for those branches.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ff3aad16851afd8e8eb894bd50c83121a142943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ff3aad16851afd8e8eb894bd50c83121a142943"}]},"branch":"refs/heads/stable/2025.2"},"fb390d60f06ca4316ea2b4e830e6b4f1f261cc58":{"kind":"REWORK","_number":2,"created":"2026-03-26 12:15:43.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/2"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 12:14:11.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb390d60f06ca4316ea2b4e830e6b4f1f261cc58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb390d60f06ca4316ea2b4e830e6b4f1f261cc58"}]},"branch":"refs/heads/stable/2025.2"},"c32903d8c524df4aa6a6a1898fb313d054a7802d":{"kind":"REWORK","_number":3,"created":"2026-03-26 12:16:33.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/3"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 12:16:25.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c32903d8c524df4aa6a6a1898fb313d054a7802d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c32903d8c524df4aa6a6a1898fb313d054a7802d"}]},"branch":"refs/heads/stable/2025.2"},"d8fd0c0e717836ebdb140ca281d074fef2d7296e":{"kind":"REWORK","_number":4,"created":"2026-03-26 14:31:44.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/4"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 14:31:37.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8fd0c0e717836ebdb140ca281d074fef2d7296e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8fd0c0e717836ebdb140ca281d074fef2d7296e"}]},"branch":"refs/heads/stable/2025.2"},"e01d90f5624680376c61fc92f4b40dd137562d21":{"kind":"REWORK","_number":5,"created":"2026-03-27 07:46:11.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/5"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-27 07:45:58.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e01d90f5624680376c61fc92f4b40dd137562d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e01d90f5624680376c61fc92f4b40dd137562d21"}]},"branch":"refs/heads/stable/2025.2"},"96854c96d2c294e59727b482ec259db78e25c2ff":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-27 07:48:07.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/32/982232/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/982232/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/982232/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/982232/6"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-27 07:47:55.000000000","tz":60},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bit more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96854c96d2c294e59727b482ec259db78e25c2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96854c96d2c294e59727b482ec259db78e25c2ff"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
