)]}'
{"id":"openstack%2Frequirements~982376","triplet_id":"openstack%2Frequirements~master~Ifbe61bd81f5610b822287b55458a3a20c0a38d3a","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-01 21:02:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbe61bd81f5610b822287b55458a3a20c0a38d3a","subject":"update constraint for horizon to new release 25.7.1","status":"MERGED","created":"2026-03-27 09:01:11.000000000","updated":"2026-04-01 21:04:31.000000000","submitted":"2026-04-01 21:02:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982376-new-release","meta_rev_id":"fa9537a19274d6d77d4436e4f09aa0f5c14d0d17","_number":982376,"virtual_id_number":982376,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 21:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-01 18:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-27 12:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-01 18:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 10:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 12:26:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-27 12:26:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-04-01 18:34:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d6ef2011cd1423fe825ebabd41af925e191e46f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-03-27 09:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecabd96318590fe7515aa8a144f2bbbacc82e53c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f75c393dd14a0f8d1144088129cb71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854be3a640014b9ba3527d2e138a629c : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b8ba0f338d45bb9fe08a848552b6f6 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71c0a33cf28b42dcbc7bbec37e8b6e09 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c67bb594ceb4418897c5a55f4f4dcce : SUCCESS in 2m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e206d3251a3c431699c61dc090b2c169 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a1bf851315045a687ca1cd3277f8444 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abe79ff02d0c44c79d93deee79af343e : SUCCESS in 2m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3f6c47f60ead454fbeddfe6617194d15 : SUCCESS in 3m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bd88f2f0643c4e1ba91066971d3bfa1e : SUCCESS in 3m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e671b00875ad424fabd46898cc6e707f : SUCCESS in 6m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fe88350dcd844b4c8cd79b02168868c8 : SUCCESS in 5m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d523d90496374ec9bb3bf1e419e2e295 : SUCCESS in 5m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/628d7e9d3e794fab8ab53146fc71bb4c : SUCCESS in 5m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/26ecd3f37a15461f83c50182fb494295 : SUCCESS in 4m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d66674e5d523430db3624b037e4c7869 : SUCCESS in 11m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8c727076866a4d76b82f9389484c0d24 : SUCCESS in 4m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2d49904a9f9844bb93f791536380e412 : SUCCESS in 5m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/396f348b9f4f4840b4f6f70733517e37 : SUCCESS in 6m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e1acfd7b86fc4c7f93cb02e44a7bc8df : SUCCESS in 14m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02adba6a92194a36b4034c12f63c9121 : SUCCESS in 6m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/130855e124c64713a3fbc902f264410f : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9c3ee7cd578a4a21b878a4d11be8b698 : SUCCESS in 9m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/001e537989634d2ebf2c127d87abc71e : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/55f0cf48e7dc47b7bfb869fd8f30e496 : SUCCESS in 10m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bf9d82d52e6740368921e6421549d7b7 : SUCCESS in 3m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0ef221b1b7b44aa9aaf98ce0b455d359 : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6d372daa337f42b1a9717d863f6e3077 : SUCCESS in 11m 53s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/72daf6532e0f426999e091fd03d0bbf2 : SUCCESS in 17m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/22ae1d55027a43998da282fe913044c2 : SUCCESS in 7m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/827c2d362ac341d7840d1200b438fad5 : SUCCESS in 15m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/aa67135c97ee493b8d796628670b15d4 : SUCCESS in 2m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71d0614b8cca4deba0551ed61305f6c8 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a98cc06bb0764637bc54d2155b8417c8 : SUCCESS in 25m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bb561704122342a48a18f7ade40b3320 : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/40c62a2754514b4b865596dfed6aa43b : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4557c7f5a178447dbf7b5202531ad63c : SUCCESS in 7m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f2c1de664e094e0c872a553b8759c130 : SUCCESS in 8m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/958d68a9b01b4cbdb806f54413c45e11 : SUCCESS in 4m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5824b22616b244299b67432b25d8c5df : SUCCESS in 1h 05m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18911b6b8d204eed99fc506f9d4c7d9c : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a72f53585c034ef79d4d6bb07199026cf723fbf5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-27 12:26:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a5a5fb5fbb9d325362d48415e2969e1448e2e00","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 18:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"906ec9679c06c6cb09990671c0f9497c069b15d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2df864878c0ab46503ce6590f68ec817dd394a1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:02:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25038ebcc4944f199208570d715bada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3adaf47b31a4aec8e30132bfb210aa9 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06def441c514de4a81edc9e12cffd1f : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20762bc46d4d4722a8e0e7d686a666a7 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad8d364db9f44b1dae22eca8f2177e7d : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41fe708600154a4297a07bcad1fc69e6 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/243f09ec9a86447d81dc7bd44b1f134e : SUCCESS in 3m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c3cac562a00d41f089bf1be2baab13eb : SUCCESS in 4m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a02ef4dad1a4b77888d4c5d438bce9d : SUCCESS in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/48177b216df74d9e9d7cd0b54c9de009 : SUCCESS in 4m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9fc8b26b057542cb96e74267408604ae : SUCCESS in 6m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/711f545b58ad47858f8f92d6351d7392 : SUCCESS in 12m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/527b39df3de04056a2796023d44091ef : SUCCESS in 5m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/97ca08adf18844d294951d02f119a3c5 : SUCCESS in 8m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5fe60606b96946cfbd0a598201b6b543 : SUCCESS in 8m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c694b8a928044f55869f4c4750758868 : SUCCESS in 20m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/138d020f9e0b4910827f723ecc01e42d : SUCCESS in 10m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/da37a48cc0a1401dabe1107fa5704426 : SUCCESS in 13m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/90711a7359864b099014640e9350a826 : SUCCESS in 12m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5d476469d8b14ca1aa40dfa5a7aa4ea3 : SUCCESS in 3m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/728c8785eea74e059e2a7feef3c330b2 : SUCCESS in 6m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ca5ef0955efa40ad9e9f11c0841c5a89 : SUCCESS in 29m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1d842171e30b4622838e8651fc7baea2 : SUCCESS in 9m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9a5a62ba1e2e495c8dd6af348939c59b : SUCCESS in 17m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3a94bc64d7414fa491d74d62cd0da3ee : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46861d8217044e228db3bfb41487c1eb : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2c88678b9f44ecb83d8d14ad112bdb2 : SUCCESS in 23m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bdc446e043d24d23a9debd8907a7b3ca : SUCCESS in 6m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2f6ea42a0d714eac807600edc745de6f : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ceda7ca126f34165b6dc00d3852665e9 : SUCCESS in 7m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4f6d73cd1454426190aab9863b0486c4 : SUCCESS in 15m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1bce2183974c4abb9c3dd79729fc86c6 : SUCCESS in 7m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e57846c365b54f08aacccce1f491d018 : SUCCESS in 1h 05m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c19aa293fe32431b8cfcb410b03312c7 : SUCCESS in 1h 41m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b1fc66b1cb963f48f5f7e9190ccf0dd7b3f51283","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:02:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa9537a19274d6d77d4436e4f09aa0f5c14d0d17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:04:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb4b7eeadee845dfa0ff59a5047c5ea6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83c5ed7a5d3415b92eb283f74f417fd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1207a2a381111df3558e08bcea3b6a652eafaf33","revisions":{"1207a2a381111df3558e08bcea3b6a652eafaf33":{"kind":"REWORK","_number":1,"created":"2026-03-27 09:01:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/982376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/982376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/982376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/982376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/982376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/982376/1"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-27 08:51:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-03-27 08:51:10.000000000","tz":0},"subject":"update constraint for horizon to new release 25.7.1","message":"update constraint for horizon to new release 25.7.1\n\nmeta: version: 25.7.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Commit: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Change-Id: Ia5858d98322480396d9e89b10ed3fee965b7d10c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ifbe61bd81f5610b822287b55458a3a20c0a38d3a\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1207a2a381111df3558e08bcea3b6a652eafaf33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1207a2a381111df3558e08bcea3b6a652eafaf33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
