)]}'
{"id":"openstack%2Frequirements~982875","triplet_id":"openstack%2Frequirements~master~I1b00882ccbe3ffc61645f8d6d3f376467b5bb1a8","project":"openstack/requirements","branch":"master","topic":"deprecate-vitrage","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-30 15:01:57.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":["deprecate-vitrage"],"change_id":"I1b00882ccbe3ffc61645f8d6d3f376467b5bb1a8","subject":"Stop vitrage requirements sync","status":"MERGED","created":"2026-03-31 16:55:41.000000000","updated":"2026-06-30 19:00:23.000000000","submitted":"2026-06-30 18:58:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982875-deprecate-vitrage","meta_rev_id":"423711fcff888028c2a3f5c44b136898a663f1a6","_number":982875,"virtual_id_number":982875,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 18:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-01 18:09:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-03 05:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-03 05:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 18:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-01 18:09:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-03 05:30:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"061016dc9028b2dde166537023b1acabdaadfb3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 16:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b36640e46c94f4cdb5a9e56cb80de24e9a15609a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:50:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4827f4a58b434b5d9734acb7d00a792f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe9204f3ef146468e1b300febea8f49 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887e214160864efcb0cacf797d029de9 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be234448d8354bf9b332295fd2235470 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faeb6f78645a438996e6410ba7f2c872 : SUCCESS in 8m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db3607e234d3474b815656a269f8f73d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62738815ff3641aeb84d4f86dc4a912c : SUCCESS in 3m 20s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/44134d3598c642568d193a89eba618d9 : FAILURE in 4m 07s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85d80018048849379741ead2d5adc2f7 : SUCCESS in 37m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25b4893e8b644e4480b89199ee29b05c : SUCCESS in 1h 49m 19s","accounts_in_message":[],"_revision_number":1},{"id":"18dcff46c6c13bd90083ae6e669d1e84bf85cd41","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 18:09:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2c54e52bc7de8153910be06719183d5c1c5045a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ca7bf7fef37b03e5dbb95c87b881a7e68f7286d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 05:30:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a15bf96d07841db90ee4a2250f831b7\n\n\nWarning:\n  Change 982869,2 in project openstack/governance does not share a change queue with 982875 in project openstack/requirements","accounts_in_message":[],"_revision_number":1},{"id":"3b24953cb89ef5fd0d10ea7285f86054a9c03139","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-30 15:01:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af40d09e3c6e6d110efe16da3627f57523e567db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 16:52:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5d38a5e6d14c2c86c2eb9ee4071918\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588eabcf3d2942e691e1fda69b67f0db : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ebb01f7e494f9e9609396fe0eda2d5 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cefc2d8b79774ce69f87b8b9da672974 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2a757d96b4a4817afececea5f2ed68a : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/daf454e65551443c93272d8d2d6c60e5 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4bed00c72314e69b49a5c5cc3f0c982 : SUCCESS in 2m 55s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/db392a37d7ab4f3fa4ad76bc36d91c66 : FAILURE in 3m 21s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/788bc173243b45d68f09888b6c387d46 : SUCCESS in 36m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5020b6d61c5461aa3884469205cb105 : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":1},{"id":"92cb6b31d74ecc559932e5273192fd3c6554bac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 16:53:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05bee42ebb1ee72bb4e6bd80913f99089824f39f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-30 17:13:16.000000000","message":"Hashtag added: deprecate-vitrage","accounts_in_message":[],"_revision_number":1},{"id":"8e5b6c06308a7a69c6bf7900a40c8923aef6a9b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 18:58:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aabb7215d4944312a1675bf58fdfa2fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e2a793554a4dd9b9ab50dd7141f643 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c997add5ca1e4efbb4c61ca754f387c5 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e0240f731ea40518abd6cecec3ba798 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a31afb994f304fd1827891ae4fc65a9c : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ac8e61135e24ef181b468c61fbf7c1d : SUCCESS in 3m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4da9eebb007f4bba854cc01d04a8e719 : SUCCESS in 35m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd61971006f4478d9d5f0fed0e047949 : SUCCESS in 1h 45m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0c6a7f373f31e4cb8616b23b806f4acb114f4ff9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 18:58:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"423711fcff888028c2a3f5c44b136898a663f1a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 19:00:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59204c5222fc414395e9cbc26487da4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe54c419f864012ad4086d915f5d32a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b81b64dc9a780b056dabc4907ac0a482446dd95","revisions":{"9b81b64dc9a780b056dabc4907ac0a482446dd95":{"kind":"REWORK","_number":1,"created":"2026-03-31 16:55:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/982875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/982875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/982875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/982875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/982875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/982875/1"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 16:55:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 16:55:21.000000000","tz":120},"subject":"Stop vitrage requirements sync","message":"Stop vitrage requirements sync\n\nWith project aimed for deprecation, we can safely stop syncing\nrequiirements for it.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/982869\nChange-Id: I1b00882ccbe3ffc61645f8d6d3f376467b5bb1a8\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b81b64dc9a780b056dabc4907ac0a482446dd95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b81b64dc9a780b056dabc4907ac0a482446dd95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
