)]}'
{"id":"openstack%2Frequirements~983036","triplet_id":"openstack%2Frequirements~master~I7c0a44ac27741f1596742d54a1746564f7afa2f9","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-12 09:57:00.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"34751":{"account":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"last_update":"2026-05-20 07:52:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7c0a44ac27741f1596742d54a1746564f7afa2f9","subject":"bump gunicorn to 25.2.0 due to deadlock in 25.1.0","status":"ABANDONED","created":"2026-04-01 12:06:13.000000000","updated":"2026-05-20 07:52:03.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b614dd0eb2f5a688f787bffd5add3aaa70b46fa","_number":983036,"virtual_id_number":983036,"owner":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-01 13:53:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-04-01 17:57:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 13:53:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-01 17:57:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5ac8ee852aa8be23351a9b674f31667c8663756b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"date":"2026-04-01 12:06:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fe1655e65f70a2579a77165e750efed16487bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"date":"2026-04-01 12:08:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ae4224610520242c7f7a80f48b5e4d1906ea3385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:53:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13547ea6c2594ae881129eedc13acd0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83a9fa25be94e8da341f2a00e628949 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e884207ef643c49d2ebc4377cf29f5 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45dc093879c2478f981cfc81211366b1 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38ac4a073e8d446a9bdec82e7b0d16ad : SUCCESS in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0bc6e45d5d7437d89c953ea318db90e : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/052d9ae2268c486e80e285ae127f5390 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0891928e19c44d31878185eee0ab9914 : SUCCESS in 4m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/735899da2c8444ecbf7b554d1d1d9eaf : SUCCESS in 3m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d3108d6a41e441f1b32345df6b7f937d : SUCCESS in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/18b3f09b77ae4205bc25c130b07d06b1 : SUCCESS in 11m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4559eebb4d314cf48a69e337ec6b26bc : SUCCESS in 8m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b3a7c80c0b614e3d8a017ccff0cb2c2c : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/38a8991918bd4b88aa69ae41b7217ece : SUCCESS in 4m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a386c94d13784de09f5aee1a86cd4924 : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/670fbcb307024ce09c9f231410a17aa4 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c610bbab511e49da968b5e7524dcbde7 : SUCCESS in 4m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/477a0d908d114078bdfcad986014b3c7 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/abda29f3d7ad44b196305a80ab6a9872 : SUCCESS in 8m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/26ebb256fef348db9834ec315385d2ab : SUCCESS in 16m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b50cfdfb94874db4b55370b7a16aa220 : SUCCESS in 9m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3ff1583695224241874e8bbc2ea034da : SUCCESS in 5m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/817c22d114a04174b83663a92b435341 : SUCCESS in 12m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/be2efad9764d4bdaa177879b1d365055 : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b83e5ae617b345678317b6bc71021d42 : SUCCESS in 9m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ff0bb6e79f2748a9b9b4c5cd613f4255 : SUCCESS in 2m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/262dcfa058ba438c9bc4747be8594c3c : SUCCESS in 3m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e5f171b9b6e54c8383129136d5bc48ad : SUCCESS in 9m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/da051023d05c477eb220c53e9a358d45 : SUCCESS in 26m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a329810e36e34aaf80f86c20a05dacbd : FAILURE in 3m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/18212300858e45f1beafd94a126900af : SUCCESS in 15m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f14cec010f2e4b90a9b822db75265bb8 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1fa128d4d83a4e91be9f5b6a0757c78b : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d388363a75ba422a97d4dd48106d3b44 : SUCCESS in 27m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bdec2be5f5e440bf8db68c5ff5b5decb : SUCCESS in 6m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c40bae7b0d134d49916d1a626a934ec0 : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8bba89783cc4634a27d349557695fdc : SUCCESS in 3m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/84a6adffffba449a8bb48562ea035948 : SUCCESS in 21m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3454422ff43e431b89285d4ca246a312 : SUCCESS in 6m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e96af66c7b394461bf7829728e9019dd : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fd169b928984900b6dd9d5bba7398c4 : SUCCESS in 1h 41m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6a484836bd6570ec97ba14b00e671bd94819ee50","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 17:57:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d675c457afe58f815d87770e95d37a07fff80fd4","author":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"date":"2026-04-02 08:23:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca916536367337671ca356a0c91b9a9833afd274","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-12 09:57:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b614dd0eb2f5a688f787bffd5add3aaa70b46fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:52:03.000000000","message":"Abandoned\n\nwe have gunicorn\u003d\u003d\u003d26.0.0 in master now","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e280eac4128a4fd256938507aa72f29749a0e62","revisions":{"044464318cd089e86339b8cac62652e1b99b4176":{"kind":"REWORK","_number":1,"created":"2026-04-01 12:06:13.000000000","uploader":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"ref":"refs/changes/36/983036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/983036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/983036/1"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"Jasper Ras","email":"jras@hostnet.nl","date":"2026-04-01 12:05:59.000000000","tz":120},"committer":{"name":"Jasper Ras","email":"jras@hostnet.nl","date":"2026-04-01 12:06:12.000000000","tz":120},"subject":"bump gunicorn to 25.2.0 due to race condition in 25.1.0","message":"bump gunicorn to 25.2.0 due to race condition in 25.1.0\n\nGunicorn 25.1.0 has a race condition [1] causing the forked worker process to\nhang indefinitely.\n\nAmphora agents install the version of gunicorn defined here and new ones\nare intermittently failing due to this deadlock.\n\n[1] https://github.com/benoitc/gunicorn/discussions/3509\n\nChange-Id: I7c0a44ac27741f1596742d54a1746564f7afa2f9\nSigned-off-by: Jasper Ras \u003cjras@hostnet.nl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/044464318cd089e86339b8cac62652e1b99b4176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/044464318cd089e86339b8cac62652e1b99b4176"}]},"branch":"refs/heads/master"},"0e280eac4128a4fd256938507aa72f29749a0e62":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-01 12:08:28.000000000","uploader":{"_account_id":34751,"name":"Jasper Ras","email":"jras@hostnet.nl","username":"jrasper"},"ref":"refs/changes/36/983036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/983036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/983036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/983036/2"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"Jasper Ras","email":"jras@hostnet.nl","date":"2026-04-01 12:05:59.000000000","tz":120},"committer":{"name":"Jasper Ras","email":"jras@hostnet.nl","date":"2026-04-01 12:08:28.000000000","tz":0},"subject":"bump gunicorn to 25.2.0 due to deadlock in 25.1.0","message":"bump gunicorn to 25.2.0 due to deadlock in 25.1.0\n\nGunicorn 25.1.0 has introduced a deadlock [1] causing the forked \nworker process to sometimes hang indefinitely.\n\nAmphora agents install the version of gunicorn defined here and new ones\nare intermittently failing due to this deadlock.\n\n[1] https://github.com/benoitc/gunicorn/discussions/3509\n\nChange-Id: I7c0a44ac27741f1596742d54a1746564f7afa2f9\nSigned-off-by: Jasper Ras \u003cjras@hostnet.nl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e280eac4128a4fd256938507aa72f29749a0e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e280eac4128a4fd256938507aa72f29749a0e62"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
