)]}'
{"id":"openstack%2Frequirements~983159","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0","project":"openstack/requirements","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-04-03 08:30:59.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-03 09:39:59.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-03 09:39:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","status":"MERGED","created":"2026-04-02 05:24:08.000000000","updated":"2026-04-03 09:39:59.000000000","submitted":"2026-04-03 09:39:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983159","meta_rev_id":"d9c9fa0d9da45279e04e310b1ac9dcc579122e2f","_number":983159,"virtual_id_number":983159,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-03 09:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-04-03 08:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-04-02 11:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-02 15:23:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-04-03 08:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 06:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 06:48:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-04-02 11:54:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-02 15:23:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-02 15:23:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"33b24ac005d79da7b57e3046f96228c25d02f34e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 05:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa98fb4d9a97ed11afd9acee6f2ff286d1ea8aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 06:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0063eb58040f454d99f274bfac2ae285\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84421b33e3ed4436ad0ab054444b231e : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2525f19201546b18c77b7095c320f14 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43869e5e880a40098eed1c096244c85e : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/101206755bf5498290d283b83910de46 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddf1ba857c47496b8843b9d463d6a319 : SUCCESS in 2m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f1fb6463d84a96b300d4d00e518590 : SUCCESS in 1h 09m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/907ba3d4e32345bf85eed051075d6367 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c14510ad7d0ec0b6ba4d6849f7f6347868fe7ab0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-04-02 06:48:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d5a0e229adedbc98b6d76c1537ba3078c913993","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 06:54:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a35851d20928a6e95fd9602d08f4dc7f74c0dcde","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-04-02 06:57:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9483c13bf257a2b9443ca1bddb1ad949fcd64ab","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 07:19:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3337987c9b5a219ac7c982de97da294e149927ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 07:20:30.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0ecc24286f2d47936229cd8fe2b794f3d37131f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 07:28:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5cd8b78279d8dbbac2e3e42ab81e91624d635fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 08:26:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85d6cc8e6e224f6e8d2986f16eec73fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979f245b8e0641ca832605b459106362 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1a31ef7d03446d9bd92c1cafa64d12 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ee1360f9f0344a0a5ff6a5dc449389b : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cc497fbcede43a68a1ff0194257b074 : FAILURE in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79d770ab6d8f4f24bfc6869fca363986 : SUCCESS in 3m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a7712db942864685a371decd6fac09d5 : FAILURE in 8m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8977b832dcd34c7eb77e62d5ed9a1aa1 : SUCCESS in 6m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1412b95253d249c395086c47e0575c83 : SUCCESS in 17m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d09bc6793a774a34853c4f33f8d002e6 : SUCCESS in 20m 07s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cc783ae9d9ff4ef88f73b99f383a92d7 : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b683a1c76ea74b9b85a392370da655c8 : SUCCESS in 5m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0320b98023f54a2e8abea3dd9cc15294 : SUCCESS in 3m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7005f516fcda4569af34f7f2bc1e8562 : SUCCESS in 8m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/47557ab7ed6745c9b1ef699ea0a6ef1f : SUCCESS in 6m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/176455425d9248b8b9c5c52ee66e2a25 : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f4324bcfd0c643d18e437ae858f3a2b4 : SUCCESS in 8m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/42053c125850478a8d955f7084db62f7 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34b68e5e7370410ab15d150a8ae5a8ee : SUCCESS in 3m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a726a62724c74230a4a74e8d533904e0 : SUCCESS in 5m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/56b2ad1ad4c84eeba89649fe89298bfa : FAILURE in 6m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d944e668c3b74a549dc7045191b02526 : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e79308137f574e8abf83a137ebbefca6 : SUCCESS in 7m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f0c0a1dbc9c642e99602e4947a86e7ba : SUCCESS in 3m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6621be778f2c4b95a403e4d32475c6c2 : FAILURE in 5m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/59539de6fef64412a57f37e275c5944d : FAILURE in 4m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8d7712bcc9ac474485f93546a77b3d7b : SUCCESS in 15m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d59c36746ba4342b4b819bad5a74a85 : SUCCESS in 7m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6287e9f91a2a4fdc8c4db5bc32868df6 : SUCCESS in 10m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3fdfaf8c09b74b2cb2e5cfcc53815982 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7205c39c729b4af18713590d5bfa3bf8 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fa55d5e8cbd45adbba0e50363977428 : SUCCESS in 25m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/38fda832260d4ec497d3b80372f97816 : SUCCESS in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3c932377a0394be2942a57d31fc81013 : SUCCESS in 2m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c4edc2d9e2d46fb9ef7e04e399730d7 : FAILURE in 4m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/98d0b94e18af41818a54a76f95cfcf48 : SUCCESS in 17m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a37b8db546b4c72a94ecf53478cc6d5 : SUCCESS in 36m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9aa4e7621bc141609cd111478cb6f1cf : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":3},{"id":"839f4a4a46eaff78685381ae7c5ae18613be2ad7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 08:32:47.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e8cf045a9822a97aa930676017b3e017f658657d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 08:54:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ab6177cac3c1c4f54c4e260da4921c5b2f1dfd5c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 08:54:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"798d6ba18d9a8850f0ddba641075757e0011da5f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 08:54:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ab033160e21b2fa83fd41e51b9cb6972eb7ff43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 09:30:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e080fe8b667e057a61f5aecbc95493f379b035cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:32:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667774809b174bfd97952c4841d28094\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f290d16ba3ea4e8ebdfaad84103900b3 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d965500781d436e86db23752299d281 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7109a73ca602463bbbf1e1112f9f762a : SUCCESS in 2m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bcddb5493ebb4c668024083e85905ee1 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a089eb789b884dd79ba6eda8e240f4ac : SUCCESS in 3m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd47d555249f4180b33f4e64389a3266 : FAILURE in 4m 50s (non-voting)\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23b3fa7862fa4965b7a918b1b9e44c4d : FAILURE in 4m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eace6fcaede64776ab04935302a43628 : SUCCESS in 56m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea7233b837ce4d1dade94692c2fa40c6 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":6},{"id":"1895a96f7feb29088f8651881cc5bdfbc5d12fed","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 11:54:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d940e0fda89a6c6b756a3a99f0b51c0ccaf5c98d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-02 15:23:13.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0a68b5c964a95f475a10bf25d5fd0bc5ce7015ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-04-03 08:30:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8238bc34109e0dc6c4e8cec5a32caeb848be8645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 08:31:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b5baac8cddedbac9c033c9ce8e377fdb5a2f4b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 09:39:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c62dbcce98346abb23f08d66daa0aad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a76a118cccc42cf9f9dee515300f239 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c540b0a94f440d4ab2c17cb47b741ec : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/685a26441a4c42bd81058b867519a6da : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb29cbcfc86546e8b206f8c541fe903a : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb0b59e0d8404706b89a7a98ad0cfcf1 : SUCCESS in 2m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43716c9daa264c18b72f3a62d2fdd597 : SUCCESS in 1h 03m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a40dd83e1a9c476ca94a1a42bedbc5d6 : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":6},{"id":"d9c9fa0d9da45279e04e310b1ac9dcc579122e2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 09:39:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dc1a4fe109658aa0d8b25575526e758bf768adc3","revisions":{"d80c795a8b17a98e39f04551cf60a3193cf756c2":{"kind":"REWORK","_number":1,"created":"2026-04-02 05:24:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/1"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:24:02.000000000","tz":120},"subject":"Add pip.sh to allowlist_externals in py39","message":"Add pip.sh to allowlist_externals in py39\n\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d80c795a8b17a98e39f04551cf60a3193cf756c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d80c795a8b17a98e39f04551cf60a3193cf756c2"}]},"branch":"refs/heads/stable/2025.1"},"a91b768af83b38917193daf55105a2fc0019f56a":{"kind":"REWORK","_number":2,"created":"2026-04-02 07:20:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/2"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 07:20:29.000000000","tz":0},"subject":"DNM: Test branch 2025.1","message":"DNM: Test branch 2025.1\n\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a91b768af83b38917193daf55105a2fc0019f56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a91b768af83b38917193daf55105a2fc0019f56a"}]},"branch":"refs/heads/stable/2025.1"},"3f23b78b5816505a32345e8ed3da14fcdaba3daf":{"kind":"REWORK","_number":3,"created":"2026-04-02 07:28:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/3"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 07:28:11.000000000","tz":0},"subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","message":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals\n\nFollowup after [1] which did not launch all jobs\nand omitted breakage in py39 job and cross jobs.\n\n[1]: I5cd04d37068daaccbd91401489918666cb1d61c1\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f23b78b5816505a32345e8ed3da14fcdaba3daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f23b78b5816505a32345e8ed3da14fcdaba3daf"}]},"branch":"refs/heads/stable/2025.1"},"6503edf2fc44482e26560241f629793554630af7":{"kind":"REWORK","_number":4,"created":"2026-04-02 08:32:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/4"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 08:32:47.000000000","tz":0},"subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","message":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals\n\nFollowup after [1] which did not launch all jobs\nand omitted breakage in py39 job and cross jobs.\n\nAlso remove check-conflicts usage from py39 that\nwas omitted in [2].\n\n[1]: I5cd04d37068daaccbd91401489918666cb1d61c1\n[2]: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6503edf2fc44482e26560241f629793554630af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6503edf2fc44482e26560241f629793554630af7"}]},"branch":"refs/heads/stable/2025.1"},"6deb696c4e066f7b4f6bcb18f760ad510d0846c3":{"kind":"REWORK","_number":5,"created":"2026-04-02 08:54:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/5"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 08:54:04.000000000","tz":0},"subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","message":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals\n\nFollowup after [1] which did not launch all jobs\nand omitted breakage in py39 job and cross jobs.\n\nAlso remove check-conflicts usage from py39 that\nwas omitted in [2].\n\nMark cross-keystone and cross-osc-tox-docs jobs\nas non voting due to failures related to\npkg_resources\n\n[1]: I5cd04d37068daaccbd91401489918666cb1d61c1\n[2]: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6deb696c4e066f7b4f6bcb18f760ad510d0846c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6deb696c4e066f7b4f6bcb18f760ad510d0846c3"}]},"branch":"refs/heads/stable/2025.1"},"dc1a4fe109658aa0d8b25575526e758bf768adc3":{"kind":"REWORK","_number":6,"created":"2026-04-02 09:30:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/983159/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/983159/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/983159/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/983159/6"}}},"commit":{"parents":[{"commit":"d69d8e81a0f99f5a7f7f65c824d97ca0882bc530","subject":"Merge \"Drop Monasca\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d69d8e81a0f99f5a7f7f65c824d97ca0882bc530"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 05:23:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 09:30:51.000000000","tz":0},"subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","message":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals\n\nFollowup after [1] which did not launch all jobs\nand omitted breakage in py39 job and cross jobs.\n\nAlso remove check-conflicts usage from py39 that\nwas omitted in [2].\n\nMark cross-keystone and cross-osc-tox-docs jobs\nas non voting due to failures related to\npkg_resources\n\n[1]: I5cd04d37068daaccbd91401489918666cb1d61c1\n[2]: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nChange-Id: I807cf2e90d48b725fa5fb1f01b83ec08ff124fa0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
