)]}'
{"id":"openstack%2Frequirements~983173","triplet_id":"openstack%2Frequirements~stable%2F2025.2~I4fef5b1f0bdb09d8e7cbec477608ebef546c8533","project":"openstack/requirements","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-02 16:40:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4fef5b1f0bdb09d8e7cbec477608ebef546c8533","subject":"DNM: test 2025.2 branch","status":"ABANDONED","created":"2026-04-02 08:55:14.000000000","updated":"2026-04-02 16:40:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2639c36121969089ff1b5f73da4dfac6faa0f2e8","_number":983173,"virtual_id_number":983173,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-02 10:48:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 10:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d62dfe483ad3bd65a23983b6a4aedd86f73893a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 08:55:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"763d7dcc47ff7ff60a069c4cf78fe93178847860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:48:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2260cb49673455598a059f646039850\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70b044507ea4065ad89340a25aafc85 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9541f09287e476cac77036213ed42c2 : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db6fa14873c142ddad3eeae758a0d950 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3da3beeb4f9f4c38baad58dd9ff0353c : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd071329034e47c08c02db6ac1bd745f : FAILURE in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f67683897a54df983f06a80c756aa5d : SUCCESS in 3m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1b8355acc8d2480cb3bdf3c3e87779cc : SUCCESS in 9m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/83351a99723d4e53942fd641a2a59db4 : SUCCESS in 7m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/21123f30bec74b13ad5c6c199a5d5a5a : SUCCESS in 18m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/489521d71720451cb4079386c24ef129 : SUCCESS in 23m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/812cdb301dcc4703b9474547a09ffb46 : SUCCESS in 5m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/811bc0ba81bc40a2b2c9cc08a734a3c2 : SUCCESS in 5m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1ee5d4faff104d87b83b177c39a4c6cc : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3352e393e1664cd6a8485bb40d145e7e : SUCCESS in 9m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c3c11daaa1ea443ab97da2e9a3167857 : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0d790c48b6c6404c8035b34fb5fd1741 : SUCCESS in 3m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/45c278119c1840faa5a4f8b6374cf315 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b1428ae41d3c4cff88cfd730519bb49f : SUCCESS in 17m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5d97619ddcf490394c3bea6ba066ef2 : SUCCESS in 6m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b1391f46e683444f96c005d30a7546fe : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2a2991374a734d8d994d05b8d9a0fdbb : SUCCESS in 8m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3cb17b93d6b643cfa2c1348f9f91f348 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/39446ae766254b89bf10795e21da4305 : SUCCESS in 7m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6332fe45967a42769d7da1877beef42c : SUCCESS in 3m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1b58bc1aff0346aab646f3c82b7edc6c : FAILURE in 3m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a66262e5efad410f907dd436ab80f963 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6e06b421041a4460985367175a0cd7d9 : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fd144e693e274578b5af2a57162c1e89 : SUCCESS in 7m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/94e92516230541598ce7785e07399b83 : SUCCESS in 13m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fc25e6ca786f4ebf8771ac88d51f52ad : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a18cd18d08624d6eb69becab2dd160e5 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebdf9bf4249949bba22da420386eaf9d : SUCCESS in 24m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0d8b1d880516479aa8a47757ab1aa3d2 : SUCCESS in 6m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d32438cf2da94ca681cc940adac964af : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b3b59a52682463f847a720a915d0f56 : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1b2317f1fe90402c92873790c0d58cc9 : SUCCESS in 14m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0137189b117948d990a43cbd8ba287e8 : SUCCESS in 6m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dc956f509de44fa8f8a795e0cfa235b : SUCCESS in 59m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/faad3fc362c94638b331b31d61c65e98 : SUCCESS in 1h 47m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2639c36121969089ff1b5f73da4dfac6faa0f2e8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 16:40:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ab30943aca967a7a5627adefdb58f48a6dee8d6","revisions":{"4ab30943aca967a7a5627adefdb58f48a6dee8d6":{"kind":"REWORK","_number":1,"created":"2026-04-02 08:55:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/983173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/983173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/983173/1"}}},"commit":{"parents":[{"commit":"56b6bc82479c2a2e5a4b72e2cde04aa20673853c","subject":"Merge \"update constraint for horizon to new release 25.5.2\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56b6bc82479c2a2e5a4b72e2cde04aa20673853c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 08:55:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 08:55:12.000000000","tz":120},"subject":"DNM: test 2025.2 branch","message":"DNM: test 2025.2 branch\n\nChange-Id: I4fef5b1f0bdb09d8e7cbec477608ebef546c8533\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ab30943aca967a7a5627adefdb58f48a6dee8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ab30943aca967a7a5627adefdb58f48a6dee8d6"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
