)]}'
{"id":"openstack%2Frequirements~983375","triplet_id":"openstack%2Frequirements~master~I97dbb2c4063a08328a8563ffd6791b7ce9eb03a4","project":"openstack/requirements","branch":"master","topic":"cross-py313","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-07 09:05:37.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I97dbb2c4063a08328a8563ffd6791b7ce9eb03a4","subject":"Bump cross jobs to use python3.13","status":"MERGED","created":"2026-04-05 09:54:49.000000000","updated":"2026-04-07 10:21:10.000000000","submitted":"2026-04-07 10:18:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983375-cross-py313","meta_rev_id":"2221b23bc26cc29dd23606450c85f155473ccc08","_number":983375,"virtual_id_number":983375,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 10:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-07 09:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-07 06:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-07 09:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 09:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-05 09:55:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 06:16:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"65a51fdcf7e8cf9c39f1aad714dafd277b0424cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-05 09:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8353e110020d548056c360ecc85f0884cebbdedd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:55:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  The nodeset \"debian-trixie\" was not found.\n\nThe problem appears in the \"requirements-cross-test\" job stanza:\n\n  job:\n      name: requirements-cross-test\n      parent: openstack-tox\n      timeout: 2400\n      description: |\n        A parent job to perform cross-repository tests.\n  \n        Inherit from this job, and add the intended project to\n        ``required-projects``.\n  ...\n\n  in \"openstack/requirements/.zuul.d/cross-jobs.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"04648e3a6c5e208c43150a5251f03466186cdb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:55:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch master.  The\nproblem was:\n\n  The nodeset \"debian-trixie\" was not found.\n\nThe problem appears in the \"requirements-cross-test\" job stanza:\n\n  job:\n      name: requirements-cross-test\n      parent: openstack-tox\n      timeout: 2400\n      description: |\n        A parent job to perform cross-repository tests.\n  \n        Inherit from this job, and add the intended project to\n        ``required-projects``.\n  ...\n\n  in \"openstack/requirements/.zuul.d/cross-jobs.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"f965c3611f37e102bc92b25ad443aefbd752f88b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-05 14:12:27.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"76d8332f4372759abb2a86afcc46942d0c8d868f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-05 14:12:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53985ff870f06983f574b0876f364672429dece0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 15:50:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93089ac0195c466a98a20caf0a310d56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e96b45de1e84c03964b0c4758d5f413 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26755a3e0d54c51b37a47aa87697382 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/047a2441ebb547b28e8883e6b09a2f3b : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df6c4334d298443b83dc361a722e169c : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b98ccfe4429246278384e1432326d1e7 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ddc99de2d5434bfcb69d547b940efd68 : SUCCESS in 3m 14s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7073de1d30e84eeaaac00f05e6714777 : SUCCESS in 4m 08s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2c9ba2ad545d456e995a40f3d3562317 : SUCCESS in 4m 42s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/30c8a33f086b44bf941e143d427da555 : SUCCESS in 3m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e83ef33264eb442984c47dca8d20a639 : SUCCESS in 10m 04s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1f2fbc0f18854244a93246e5f146f3cc : SUCCESS in 6m 08s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/711880ecc07c405c9e53d08a5cf0c0e0 : SUCCESS in 4m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/261d38e0f09f41edac355463c045e68c : SUCCESS in 6m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/9932e919cdfd4ad7a39d3604c609f8cd : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43f1bc885e7f421fb83372f36d2606a1 : SUCCESS in 9m 46s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/de05b86d49bd4253917b546f4befb108 : SUCCESS in 7m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9bc488c11543429586507e886d7f8be7 : SUCCESS in 10m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9eb1ab31518a4b9b9d0af0b574c6d668 : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5878c1e3ac694499b8968ebdc988d114 : SUCCESS in 7m 29s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a06e83b9613c458c99448c4e48a2c1c0 : SUCCESS in 4m 14s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/9f6c98093ef44248a70cea3f117ec78e : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f83cd717434b47a6bdf23884fd152d52 : SUCCESS in 10m 08s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/90f71464a16d4ecebbabefecdffb265d : SUCCESS in 28m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/da68e9d5a91f4949861ab87cf9eec291 : SUCCESS in 6m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/69a5ea441ee2425a9f8fc54edef75b6b : SUCCESS in 13m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/66ed48a750b14a32917ced01220b5943 : SUCCESS in 3m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce5ee5d1811e47a4baba6a00951cddba : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/528425825e224f999ff2b4715c0bb4dc : SUCCESS in 25m 07s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d8b6da882d444dd38b3f172e7fab102c : SUCCESS in 5m 27s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5596327bb0d34885ace1846a268141ad : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7728c54342e84290b51ad2aadadc3d16 : SUCCESS in 6m 34s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f27f59f0002d4b9699e7a44291dd372f : SUCCESS in 16m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/13a5c51256fe4f5f84115b5dee2135cf : SUCCESS in 5m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11792125e9aa41f7894e90198e886a58 : SUCCESS in 44m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/289d26035d2e473e8e7ae04deabd29ad : SUCCESS in 1h 36m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ddc951457d308ddf9dbfe8bd7700476ae6092817","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 06:16:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9c57f135ad14ab2b0861a5b4cd24b41788662e9b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-07 09:05:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6125103ab2f439d272ed6ef5044065c7ae6df348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:06:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96003293914123059a57301aa9f25454da812131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:18:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4d69c22dc7453e883ce4e601b3131b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc75417e713349e4a9a72862db0c7fb0 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7a5968550f46e49f5161b7aed33e73 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d88002d289a24e7ca94e7f463712af37 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea170fd0182348a9aee80e7bc4a5505c : SUCCESS in 2m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0afe2dba000d480ab30fa4ef88bd0f42 : SUCCESS in 2m 24s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ce4c4b13d6a44d239539bb09223f1b1c : SUCCESS in 6m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/25bec06358d74a5fb21d03e065647ca9 : SUCCESS in 4m 10s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/06de0f14c403434e8a3fc685803f472f : SUCCESS in 6m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ddb7d4399272423f9f539a15d744c03f : SUCCESS in 7m 34s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/aa4edbdde0fd40228d95f8c1edabeb6a : SUCCESS in 14m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f13aa9d78b5946ecba9d7c69651d957f : SUCCESS in 6m 18s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c19533a5011240859ed8488958fd4fb6 : SUCCESS in 8m 06s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dca3683b6f2d4d0da1e693a0fd8919e4 : SUCCESS in 7m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a82ff0b0d2634d8eabf251e46cca5b7b : SUCCESS in 2m 12s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d600e5f09edb446d8bb512237150426d : SUCCESS in 7m 47s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/198ee8f7e4f040cca75f0bddc854e079 : SUCCESS in 16m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6eebdfa455cc4dd190174fab2be6bc62 : SUCCESS in 5m 23s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a445873eac864b079eba188f2cbf1995 : SUCCESS in 13m 46s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/857f3629c7324c7f8d2a0f5827c5b20e : SUCCESS in 2m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e92f30b5dd7647e083f01a46c2dc1cfd : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e80ff8b1e1643e08029ca077fd8d664 : SUCCESS in 24m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/8bad27776ba04b4eb4a09daf112ba6d9 : SUCCESS in 11m 53s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/215a04f44b8d4f569c8ae352f5ca6ffb : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b307ac4d19fa484482141e9df0c800a8 : SUCCESS in 6m 09s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a936ca376abc47ea9d3b02308c6cf5bb : SUCCESS in 17m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cbf20b2ebb504a60b8973a5ec7570fc2 : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99ff9e56f7324ca98d2bf1c1f941f6f9 : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4112711130942c5a113f3cf632dd079 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a52bed2a04479b7d1ef1eb399b75f9e944e43510","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:18:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2221b23bc26cc29dd23606450c85f155473ccc08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:21:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185775f4863b44ceba4a2f3da7461501\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c016aa647cf4527a39ae378ecb21227 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fda430c4650f95765d5a5520b1e110434bff5062","revisions":{"1c49c9fff6223cd0d62d72b6be5229c701e61489":{"kind":"REWORK","_number":1,"created":"2026-04-05 09:54:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/75/983375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/983375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/983375/1"}}},"commit":{"parents":[{"commit":"b763c1f6e124482045dd2140cb2c1a698b0e9524","subject":"Merge \"Replace obsolete PCRE packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b763c1f6e124482045dd2140cb2c1a698b0e9524"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-05 09:52:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-05 09:52:37.000000000","tz":120},"subject":"Bump cross jobs to use python3.13","message":"Bump cross jobs to use python3.13\n\nThe cross jobs should use the latest available python version, which is\nPython 3.13 currently.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I97dbb2c4063a08328a8563ffd6791b7ce9eb03a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c49c9fff6223cd0d62d72b6be5229c701e61489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c49c9fff6223cd0d62d72b6be5229c701e61489"}]},"branch":"refs/heads/master"},"fda430c4650f95765d5a5520b1e110434bff5062":{"kind":"REWORK","_number":2,"created":"2026-04-05 14:12:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/75/983375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/983375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/983375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/983375/2"}}},"commit":{"parents":[{"commit":"b763c1f6e124482045dd2140cb2c1a698b0e9524","subject":"Merge \"Replace obsolete PCRE packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b763c1f6e124482045dd2140cb2c1a698b0e9524"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-05 09:52:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-05 14:12:26.000000000","tz":0},"subject":"Bump cross jobs to use python3.13","message":"Bump cross jobs to use python3.13\n\nThe cross jobs should use the latest available python version, which is\nPython 3.13 currently.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I97dbb2c4063a08328a8563ffd6791b7ce9eb03a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fda430c4650f95765d5a5520b1e110434bff5062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fda430c4650f95765d5a5520b1e110434bff5062"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
