)]}'
{"id":"openstack%2Frequirements~983518","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I5cd04d37068daaccbd91401489918666cb1d61c1","project":"openstack/requirements","branch":"stable/2024.2","topic":"setuptools-build-constraints-2025.2-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-22 14:27:52.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-22 14:27:52.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-04-22 14:27:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cd04d37068daaccbd91401489918666cb1d61c1","subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","status":"MERGED","created":"2026-04-07 08:43:30.000000000","updated":"2026-04-22 14:27:52.000000000","submitted":"2026-04-22 14:27:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983518-setuptools-build-constraints-2025.2-stable/2025.1-stable/2024.2","cherry_pick_of_change":982268,"cherry_pick_of_patch_set":4,"meta_rev_id":"55ebb79fa338d574188dc2d52ebc147b80f615cc","_number":983518,"virtual_id_number":983518,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 14:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-22 12:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-22 12:43:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 08:43:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2026-04-07 10:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 09:48:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2026-04-22 12:43:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b0cce1723c28a3515b29836b58806303c7081d26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 08:43:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.\n\nThe following files contain Git conflicts:\n* tox.ini","accounts_in_message":[],"_revision_number":1},{"id":"24e93d0bbcdf522bbb4607bc0fe67ed27c177ed4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 08:44:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"55e131a2d25cfe04e8f285431655061ab5cf7c07","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 08:46:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"728a5a0edf8df187925c386890df0242987d0044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:17:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c318829b33b4f1aad9039122940f0b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab752b01e98d44eb80c4c469079ef66b : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/396c0676b9934cc4951e40ae9b1c146c : FAILURE in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d59a50e6153b461784d68028b1956786 : FAILURE in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ed045503b1548f6aa4481fa3666e6b4 : FAILURE in 3m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0178de8b2194a5abaabea1a39ad0965 : FAILURE in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a82353b933640bab5301e2c6e85348f : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d23e72191e53469ca045f73866a554ff : SUCCESS in 1h 07m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/971160c7f2a44b9aa8c979de15980582 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ac2ad85c0614b2940216fd830b7cfb05c53bde2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 08:31:21.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f812b3bb3d61a658a32140402e8fe54c065bd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:17:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfcb8c58e42e4b359ab7ed3d66f8d54d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6861de7eaa9148c48bafa960a049a453 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca6d0552879243b1a746463388af0f30 : FAILURE in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/401aa6600f2647d6b0f3a5e4c4ec5686 : FAILURE in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7b0d1cb7749406e9f5621ee4183aff5 : FAILURE in 3m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98accfd76b0349a9a7e615a47d5a0525 : FAILURE in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f070857d9abf44ceb931785b22fddb43 : SUCCESS in 3m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3451086bffd74bf1846f2b6a979f27db : SUCCESS in 1h 11m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e4398081aa44bb093002c0ce808f9dd : SUCCESS in 1h 39m 51s","accounts_in_message":[],"_revision_number":4},{"id":"37797ae0abadd1e5d5da2d9e819c061a6c1b7061","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-22 09:48:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9acead596262f19bcdc97d8c29c39b21d3459e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:58:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86716d4fbcab495990832ff0d8e6a210\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca879094e2bd44b98b19a036fd23935e : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7547d4cebd354c5582387ea4e2d994c7 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5228ad5eb4ff4b27abaa84004286a33e : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4519b0949a75471fbe62c4e53205348a : FAILURE in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/130d362bdc7248379873127e6060a291 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f27f6215ee8e42578ac211009d0adc85 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d093a1f3824b4dde8ebb74f60d1abbf2 : SUCCESS in 44m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9678191f33954d72820f799f8c49245f : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":5},{"id":"aa71ad5a856474c9321fda19151f05e47b963ff8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-22 12:43:30.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"32c18b60e653d127e5616bb27692d237e5481348","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 12:43:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"150f2a2d02f88e13b1e8748b2edd3ce97053e5e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 12:44:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db171985e2d5b1cf1593525d6f09a62ea3d63a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 14:27:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b188dd0d708448c7bd44fd4490d72900\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ef87e50eb14344a9a1672abb9ce836 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64031edcb440412cb9601e935541b1b5 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0262827726a4b1dbebfcfcffb87e9c7 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/235447617ab44609bd5b5a48d150d750 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/164104d862d04f99a465bfa8e0cd5936 : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d96e5789ad4148b790ca58d2c94fb66c : SUCCESS in 55m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6392e2c000304e52804dabd23089cbeb : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":5},{"id":"55ebb79fa338d574188dc2d52ebc147b80f615cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 14:27:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"896dd1d077db0cb1669d8ff700b2ec024f5004c3","revisions":{"cea072f39bd8d9af68942cdd826d0c5f3d2bed42":{"kind":"REWORK","_number":1,"created":"2026-04-07 08:43:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/983518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/983518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/983518/1"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 08:43:30.000000000","tz":0},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n(cherry picked from commit 277424d1e7d5a9b0b7f2bdb126d4797232686322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cea072f39bd8d9af68942cdd826d0c5f3d2bed42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cea072f39bd8d9af68942cdd826d0c5f3d2bed42"}]},"branch":"refs/heads/stable/2024.2"},"d49cc8e33c83d9c8b35978835327dd6a05a9b0e3":{"kind":"REWORK","_number":2,"created":"2026-04-07 08:44:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/983518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/983518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/983518/2"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 08:44:42.000000000","tz":0},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n(cherry picked from commit 277424d1e7d5a9b0b7f2bdb126d4797232686322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d49cc8e33c83d9c8b35978835327dd6a05a9b0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d49cc8e33c83d9c8b35978835327dd6a05a9b0e3"}]},"branch":"refs/heads/stable/2024.2"},"22dfb0a32194b7f246d40405974a1964fdf402c3":{"kind":"REWORK","_number":3,"created":"2026-04-07 08:46:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/983518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/983518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/983518/3"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 08:46:22.000000000","tz":0},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n(cherry picked from commit 277424d1e7d5a9b0b7f2bdb126d4797232686322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22dfb0a32194b7f246d40405974a1964fdf402c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22dfb0a32194b7f246d40405974a1964fdf402c3"}]},"branch":"refs/heads/stable/2024.2"},"3b227f17bdd7947ab7191b3a590650fc0a6687ff":{"kind":"REWORK","_number":4,"created":"2026-04-09 08:31:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/983518/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/983518/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/983518/4"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-09 08:31:21.000000000","tz":0},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n(cherry picked from commit 277424d1e7d5a9b0b7f2bdb126d4797232686322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b227f17bdd7947ab7191b3a590650fc0a6687ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b227f17bdd7947ab7191b3a590650fc0a6687ff"}]},"branch":"refs/heads/stable/2024.2"},"896dd1d077db0cb1669d8ff700b2ec024f5004c3":{"kind":"REWORK","_number":5,"created":"2026-04-22 09:48:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/18/983518/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/983518/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/983518/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/983518/5"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-03-26 09:42:03.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-22 09:47:22.000000000","tz":120},"subject":"Add build-time constraints on setuptools\u003c82 to check-uc jobs","message":"Add build-time constraints on setuptools\u003c82 to check-uc jobs\n\nMany packages in stable branches depend on setuptools\u003c82 at build time,\nbecause of the pkg_resources deprecation. We can\u0027t fix this by upgrading\nthe packages like we did with the main branch, because we want to keep\nthe branches stable. So we need to pin the setuptools version. However,\nit\u0027s a build-time dependency, so it\u0027s a bid more tricky. We have to\nforce a version of pip that supports that option.\n\nNOTE(elod.illes): this patch also adds virtualenv capping in tox as\nthat is also needed for unit test jobs to pass.\n\nChange-Id: I5cd04d37068daaccbd91401489918666cb1d61c1\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n(cherry picked from commit 277424d1e7d5a9b0b7f2bdb126d4797232686322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/896dd1d077db0cb1669d8ff700b2ec024f5004c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/896dd1d077db0cb1669d8ff700b2ec024f5004c3"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
