)]}'
{"id":"openstack%2Frequirements~983673","triplet_id":"openstack%2Frequirements~stable%2F2026.1~Ie468f51dda4a458e479ffcbafd190aebd74c948c","project":"openstack/requirements","branch":"stable/2026.1","topic":"new-release","hashtags":[],"change_id":"Ie468f51dda4a458e479ffcbafd190aebd74c948c","subject":"update constraint for oslo.utils to new release 10.0.1","status":"MERGED","created":"2026-04-08 07:22:07.000000000","updated":"2026-04-08 12:23:48.000000000","submitted":"2026-04-08 12:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983673-new-release","meta_rev_id":"bc229fc14486c785f3bdf6a53c56d42e3c06dbe7","_number":983673,"virtual_id_number":983673,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 12:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-08 09:56:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-08 09:56:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 09:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 09:56:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"02b3a8a8135901b1238098469f17ab3b95b8024a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-08 07:22:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7796913263e8cc87077befaff4ebe3e3926c4a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:13:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c785cb6aff4826ad44856616d86d14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aefc48be666e4fc0839dbbc13f6989ec : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a296d3c086fd4bcd991eb4f917e385d9 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c2568f8d0b949768507b00914a2d6a5 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eae6060daac44c1bbc6534550dda310 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0172e09774b4f8bbb4773d4f2f2bca6 : SUCCESS in 3m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f30fd2dd12244bbbb4f0677fee8a1f21 : SUCCESS in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a408c6cf4445481c9dafa745e5924931 : SUCCESS in 9m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/996563dedad1427d81fec41c752d6539 : SUCCESS in 3m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4c8a162400344922ae3aff46b9e9c72a : SUCCESS in 3m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c33afe3573114943897ecac0ce6612ab : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/647510066a56419eb56e1c2aa32f3f67 : SUCCESS in 5m 02s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/384ec9f030754950a27cfeb0566382b3 : SUCCESS in 3m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9366e33b65ba4ef5ac262a40564ea9fc : SUCCESS in 11m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/33c1123fdc2640b39f9ae1f7a0e9e98f : SUCCESS in 3m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/62c74059ecd74d08ad2728ad18e2ae73 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6fae476622d640918f0de81eebe25560 : SUCCESS in 7m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0baff05db6f74bc09ead6414081c9083 : SUCCESS in 14m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/503c4a0f22234deca20e938520b50e67 : SUCCESS in 5m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/549bf24b78134e59bd99436e10d1fe13 : SUCCESS in 4m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a985a158b75b4e7980737986970b32ad : SUCCESS in 11m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/477e64c0852248678560ceafcf841a98 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0a8cf71151494498950cdba11cb53389 : SUCCESS in 6m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ff58ff744f7f4be78d0a5708c5d0f871 : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/32655e2bd69a4b8a865bc6e4cf2a9761 : SUCCESS in 3m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a4b4bce3423149b6a214fd8cdcfea859 : SUCCESS in 12m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f725184ce4804af0937ea88e641cd40f : SUCCESS in 28m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/63235232382c4a07b3e15cae2b4d3f99 : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/133b3eae71ee4ab4b9a7fcdc3c6119ca : SUCCESS in 12m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f3cfbb90756c4fcdb3cb50de4b761885 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9a14bb9cb624c0a8dc5200e6b11ca09 : SUCCESS in 2m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc0a79a5fce74a718237bedbb9398652 : SUCCESS in 28m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e1bdd54324f5492cb410fa24b031ab89 : SUCCESS in 3m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/10c0514c235c42e6ab49548d3207d49c : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29031c273bcb4e68825443efb5f6711c : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/76dbe18af39349a88cad7f76ffd0f248 : SUCCESS in 19m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e7c6581b017746798cf148adc3f1a8ed : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76b1628101084e22abe37580f304cfeb : SUCCESS in 37m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e5d6bd027b04aa49c70bd0fe2d07c7c : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e3fc9f79bad390422466cda6515a14ed5110fca6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-08 09:56:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e1432b70d77a59a751b4d1e737b290f22d47b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:57:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f72cbfba8eecc2ed7c338d36fdbc7fdf9e3a78ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:23:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7f974c2205448faff6a75df185a62b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3203dca77ea46bd8adcb170230ef3ee : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5abaf339081d4822bdb30195767eb330 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec1d6f9e812e40d5a5344152e77b7555 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a5ba5dbe13b4074bb31b39cd58fdf7d : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/765cf5bf422f4c4e8cbe6e29b378ec5b : SUCCESS in 3m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f34283cb234c4347aae319d72ed4fe51 : SUCCESS in 5m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e7c762b159194fee8c6086ab97ee7c8f : SUCCESS in 11m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3cfe21ff02c7498c8e5830074df7b369 : SUCCESS in 7m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67d0aec6373f47fb9ddc78f02d6feec4 : SUCCESS in 4m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2b4f45fb8a2a43abb5137b063adb429c : SUCCESS in 12m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/801d09d1ad094137a4c95415a702a055 : SUCCESS in 5m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d6a3fd96e6db4571a4e8d25d9b267745 : SUCCESS in 3m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4e1310dfab0a4301ac7ef766c5152aa3 : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/becc4841d40842638ceab6d546807acb : SUCCESS in 17m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93aaa341c86f445ab40753992f2171de : SUCCESS in 7m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8a6353ec17fc44f6baefbfa008d5b0b5 : SUCCESS in 7m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0547d100a1d44369964573a79d8050b0 : SUCCESS in 13m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6421f31ecb474c9ba2481e3aabff67f9 : SUCCESS in 2m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c46b8f8604724dbf8007508804c832ae : SUCCESS in 13m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/65826ae4a51f493ab8e070af2d57474e : SUCCESS in 26m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f36ba30794045baab50792653c5e76e : SUCCESS in 5m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/34d2f78e21ea4ad8a0d229364e602ed8 : SUCCESS in 17m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/07a7f76f1f2047c1aacac574d5a3d647 : SUCCESS in 7m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42de8e4c2d10458f8c0d290103e005eb : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9d93db9d6b840ddbddfb10a4d2f15c3 : SUCCESS in 28m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4a340e3ab1d7434db8ac3a5e82476548 : SUCCESS in 7m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/14e339c0cd5b4e85a9ce4c9fa9c17766 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d21eb7ed220e4f57abf30adb1da624f9 : SUCCESS in 6m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f49196d27da445d7bb52122f7d44ba15 : SUCCESS in 19m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0aee7e0b062e4083bc2e7e31ffe7eb55 : SUCCESS in 6m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdf8c0140b8947518e1a7ecfc66b7760 : SUCCESS in 1h 03m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/520cea1910794b2e8852df4fe660f783 : SUCCESS in 1h 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bc229fc14486c785f3bdf6a53c56d42e3c06dbe7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:23:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a493d10b298c35543107105a9d24be1423848e55","revisions":{"a493d10b298c35543107105a9d24be1423848e55":{"kind":"REWORK","_number":1,"created":"2026-04-08 07:22:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/983673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/983673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/983673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/983673/1"}}},"commit":{"parents":[{"commit":"6c594746204f845eccc442c02e5a2350e7e4590c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c594746204f845eccc442c02e5a2350e7e4590c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-08 07:12:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-08 07:12:06.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 10.0.1","message":"update constraint for oslo.utils to new release 10.0.1\n\nmeta: version: 10.0.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Id58d10f13c92802d2286cf30164c4d725c3e8331\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: Ie468f51dda4a458e479ffcbafd190aebd74c948c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a493d10b298c35543107105a9d24be1423848e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a493d10b298c35543107105a9d24be1423848e55"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
