)]}'
{"id":"openstack%2Frequirements~983674","triplet_id":"openstack%2Frequirements~master~Id4b40f1a886a25cfc32c7aa6f8aa748d43453b1f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id4b40f1a886a25cfc32c7aa6f8aa748d43453b1f","subject":"update constraint for etcd3gw to new release 2.6.0","status":"MERGED","created":"2026-04-08 07:22:53.000000000","updated":"2026-04-08 12:25:56.000000000","submitted":"2026-04-08 12:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983674-new-release","meta_rev_id":"3fd6d70c02b667ef4d67def72b11600f9e67e053","_number":983674,"virtual_id_number":983674,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 12:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-08 09:57:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-08 09:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 09:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 09:57:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2a5d4cc0e50f2e3b610953a527a6fc19a71a7527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-08 07:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44220e63b20b572ff55a3a6ee0de36a5e9b4767e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:16:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/652cb3ae4a384e8eb367a9b2687f0d30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28984d8401a943adbf96a20aa79fa1d8 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e463eaf48c6044d38dfc1badf190364b : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df88683e23264037a0610e9a1f2a161d : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2420f2335b0e4584b1911b5f639115e8 : SUCCESS in 2m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3612209d0d0463090eb626c2a96c7b6 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac0dc3e09ad848b58c7a9f0204bd9c7e : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a28e1fab44c4bdd89e7fdec9dbb8b24 : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ba0c0af266ed4531be14b5324df2d9b4 : SUCCESS in 7m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7982256b8abe45c7a429b97f1e7d6318 : SUCCESS in 5m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f67d4a8e2f284a86a599934753e3c816 : SUCCESS in 3m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/da45c8a0334d47689261f61e147a2354 : SUCCESS in 7m 51s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/bf5fc00321784b478d09c644a5ae06cb : SUCCESS in 4m 17s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8f266a567b64475ba19222f22edd5a4f : SUCCESS in 5m 32s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8df1d4ca9d734a8a802b4854f8a642d7 : SUCCESS in 4m 31s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0cba545c518c45f98f5578071f07a364 : SUCCESS in 11m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/315c12e73e144df1b42802213d62db10 : SUCCESS in 3m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/153523d0f5114184a2b1e7df7c04a0d5 : SUCCESS in 7m 06s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/19ba2d7a5ffd40349b7e571b137992aa : SUCCESS in 5m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/89780163b6ca42fb97659929107ae186 : SUCCESS in 15m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6ada3817f10418e89fd4e3728100152 : SUCCESS in 6m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e5e449dfed744c4a89d75aef5dcbacb7 : SUCCESS in 6m 12s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f485f571848443d7a75b965f546804e9 : SUCCESS in 6m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d54e3327c9a942b5ad28d8be922cbdf6 : SUCCESS in 3m 40s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0640a2d2b40b4217b589415e15b0ae30 : SUCCESS in 9m 04s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/6e38f084d2474ac894108914438b4451 : SUCCESS in 2m 26s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e1718362f3ae473dad76f0a2f3cc8dfe : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f5dc034cf0ea4b1baa990ebc674ce929 : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2ea5bf40226c4c2db4ec1e094ff2941f : SUCCESS in 18m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/29d482f5fbd941e184dced7d114e59ae : SUCCESS in 6m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a97fe12679e349b5b691eef94c666049 : SUCCESS in 10m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c05524ffde9e42b288309dd0517f9760 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3906b36520d498789968558a5605658 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2aa87b5c1d504c21aab726e7b74b514a : SUCCESS in 26m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e4edd89515214fa897ca98fc9f0c8268 : SUCCESS in 5m 32s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/741eba6bea614c4db49c2dea4681d0ff : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6314c6a7caeb480c9634a7dca9888fe4 : SUCCESS in 6m 12s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f8ddfb62acc1473db195b907468c2048 : SUCCESS in 17m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9570f08b38314c958bf75a0ef35c4614 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fbed5bc42ae49c79641f18757c900da : SUCCESS in 1h 12m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94ddbdfe1b744a269b47e0bd8ab4ec8c : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6699377d678faf058fefb2b5d1e43216571360b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-08 09:57:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfa3f022e091a322255afd76f814fdf855272bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23eedbacfbc8ae77b0dd79ef03fd02eb0af4783f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:25:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e55507c0e5ce453b8fbd986f69e2f3d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0b09f1498e4d518e0b022de1f133a1 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a289d761b54149169aa8959b4eed9267 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e59c4c783b094a9ba2fe181191016e06 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdb3e4458720498e9119166472d22d30 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c091661347394e9dad24c741225369f8 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e90c99e0aac46da884352c14a828867 : SUCCESS in 3m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd0919ab92594952bdb16c814fb9698c : SUCCESS in 5m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/42bd195852414a409af0f3554cd317ab : SUCCESS in 6m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/816c3774ea7d4be78904e2e4f0aac7af : SUCCESS in 6m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/65ae7f8656564ba3a864dc40d82566b7 : SUCCESS in 5m 50s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9277689bb6184718a8d8f3ff6a3f7772 : SUCCESS in 12m 28s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/78164e3236eb498f9d0e25d99d479aec : SUCCESS in 5m 17s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/07928915af314a078428171735125460 : SUCCESS in 4m 03s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/6eafcfce7ef44cd6a0e0486ee5573997 : SUCCESS in 8m 07s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cedb021f4a064644bc0bafbf4a623ff3 : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3279a63b3d148d88c417c3f20d3e903 : SUCCESS in 7m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/46f60b38d1d64705af75dda340499dc0 : SUCCESS in 7m 44s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/5f9ef81d6cab4752a5b491f43d559640 : SUCCESS in 8m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a82b4db6cea649abbcfbcf9b637c6caa : SUCCESS in 5m 12s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/cea2395a78b347fda1e105058dc02edd : SUCCESS in 6m 30s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/99e30a9bc8694483a7ddc12763c14d2d : SUCCESS in 23m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0307113cf82e49fb9f78668ec36cc8db : SUCCESS in 8m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/21c1bff6dbdf4a1398d6ae73b7d19852 : SUCCESS in 14m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2314c0ea7f464ac680e78b16bfe87d16 : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0883449abae9436da9b5c87f2d17780a : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5f169be0d4141f38a07e77e451bfaab : SUCCESS in 26m 30s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b7cd03cb27e24716a7db059654299965 : SUCCESS in 6m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b84f31fdc5264cdd8dacdb01c4851ddf : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c45fbcfeaf04cf4b3b40f3d78841b41 : SUCCESS in 6m 44s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7b8129d56f984a708994ca52619d4204 : SUCCESS in 17m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/35e2926911094560972f92150ed9e334 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8238c900dd5645978f2d8edab02cc640 : SUCCESS in 35m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c970e85b1af8422c8af63893ba52141c : SUCCESS in 1h 48m 27s","accounts_in_message":[],"_revision_number":1},{"id":"69e7ec0a45dd67ef5cadd815c6898f09e25abe65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:25:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3fd6d70c02b667ef4d67def72b11600f9e67e053","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:25:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6a96c82d2946d993e3aebcfed34f95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa03abf6c104166aa85ebc724badbdf : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9411da7123a104af44b423fd3a8c3d799fc14cdf","revisions":{"9411da7123a104af44b423fd3a8c3d799fc14cdf":{"kind":"REWORK","_number":1,"created":"2026-04-08 07:22:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/983674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/983674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/983674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/983674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/983674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/983674/1"}}},"commit":{"parents":[{"commit":"90c07b0db98601571b2c58e7dcc35853c4358d02","subject":"Merge \"Bump testtools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90c07b0db98601571b2c58e7dcc35853c4358d02"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-08 07:12:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-08 07:12:51.000000000","tz":0},"subject":"update constraint for etcd3gw to new release 2.6.0","message":"update constraint for etcd3gw to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I6c89d0b7eadfe12d1910d785da3ffed2118823be\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: Id4b40f1a886a25cfc32c7aa6f8aa748d43453b1f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9411da7123a104af44b423fd3a8c3d799fc14cdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9411da7123a104af44b423fd3a8c3d799fc14cdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
