)]}'
{"id":"openstack%2Frequirements~983708","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I9649a84ea0ef1621b7c248df160f349df5d8099b","project":"openstack/requirements","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-09 06:17:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9649a84ea0ef1621b7c248df160f349df5d8099b","subject":"stable-only: Fix jobs failing in gate","status":"MERGED","created":"2026-04-08 12:53:04.000000000","updated":"2026-04-09 06:17:40.000000000","submitted":"2026-04-09 06:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983708","meta_rev_id":"d57c208b46050108103553f99019cd824fd0aaaa","_number":983708,"virtual_id_number":983708,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 06:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-09 04:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-04-09 04:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 13:37:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-08 14:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 19:14:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-04-09 04:49:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"53c15e0437a74eb0b9d23c55745cd4b810e5559e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-08 12:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cf8affabbb74622e5eb2eb92be3569c2d146917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 14:25:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33fae2edce3c4dbead4a29a69f297eeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e756e3ae5e4c5ead64f893bdd46b1c : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/935232bcecd44e71aa68a5b9ae1cb469 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7f97458a7dd4871acd35e458c76ebe4 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1710cfc4b62e4abe88605d199ee85ce5 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fccc21e6424749bb84d4365d5dd68470 : FAILURE in 19m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44f469fc9871447bbe479b984ed9dadc : SUCCESS in 1h 30m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1ceea0526f27136365513fbe50ee073f1cd11e22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-08 19:14:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9eb86c87baeeae54f521ba2e9ef03c933607af5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 21:13:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa816a09c504cc3b5ba5d51b5178ec9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b20e3a51c0b443ba59792950e967ff1 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75b69311f45c49118b5bf341c254204b : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09e49b0131fd47a6b5ab2e948ba7d0f6 : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d45dcd915a64ea8850a5751c278ce08 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fb55e320b6a458a99031c66ab51c3a8 : SUCCESS in 53m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/863566316f754c70a5c36dcf6db3ca32 : SUCCESS in 1h 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"218970d66acb005458a98bb8c989fe4ebfaa12ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-09 04:49:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4893c0d85091d490f0bb028bf5a015c38da62b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 04:50:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"793512e5f86f0380f2d0722ccc9035c42685c26c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:17:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f0ccd013df44c9a2ac186b6a83633f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9ae10506fd4faf978d5a3f1f0e0b96 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e2f8db21b2443d85f75c4fa644cdd8 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e737545e50a4be383b0da7f76ed2d4a : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c982e68915c14d6cb58f9fccd87c677c : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11532b20da6c466ea497aea4b468ee1f : SUCCESS in 1h 03m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11c1036d986d4eab9fba2f6dbb48de89 : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d57c208b46050108103553f99019cd824fd0aaaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:17:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"283ad08ce28f960c0404f18812a7f85e81b00189","revisions":{"283ad08ce28f960c0404f18812a7f85e81b00189":{"kind":"REWORK","_number":1,"created":"2026-04-08 12:53:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/08/983708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/983708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/983708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/983708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/983708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/983708/1"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-08 12:52:01.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-08 12:52:01.000000000","tz":120},"subject":"stable-only: Fix jobs failing in gate","message":"stable-only: Fix jobs failing in gate\n\nIn [0] two jobs were made non-voting in the check pipeline, but it was\nmissed to exclude them from gating.\n\n[0] https://review.opendev.org/c/openstack/requirements/+/983159\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I9649a84ea0ef1621b7c248df160f349df5d8099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/283ad08ce28f960c0404f18812a7f85e81b00189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/283ad08ce28f960c0404f18812a7f85e81b00189"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
