)]}'
{"id":"openstack%2Frequirements~984106","triplet_id":"openstack%2Frequirements~master~I9f5f9107f79422d77f246eb45ff81444ea3a1de1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9f5f9107f79422d77f246eb45ff81444ea3a1de1","subject":"update constraint for openstack-release-test to new release 8.4.0","status":"MERGED","created":"2026-04-10 18:18:25.000000000","updated":"2026-04-11 11:59:37.000000000","submitted":"2026-04-11 11:58:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984106-new-release","meta_rev_id":"585eaa73905aa9657f128b45188c567ee1d40e7a","_number":984106,"virtual_id_number":984106,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 11:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-11 10:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-11 10:02:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 20:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 10:02:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9ee69fb6c72dbc032b28193a86ec7da89c5cf882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-10 18:18:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90d7684de0ec3f5b7bf1d0ff7625c42d0f9400ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:33:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6712001a07824eb58d2c415176999315\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd67afae8dc3445e97f463ae2d93ad0f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a0d59728cb455893c727974f4db26b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba270d552f214b949d243f6893bdb473 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fee89354f764d3fb48ca54cfbac9e6c : SUCCESS in 6m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2353859bef30427bb3789366ba64917d : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1895607e62e54935b6915f60f534ac0c : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b363fa9242a426e8d80529ab365f0bb : SUCCESS in 4m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/27dedd355b5b4d20846ab5e6ee2bf36d : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f8448b17efab49f3aa352369d4b9e504 : SUCCESS in 5m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/430c8b07bd804fad8ee9cd8f9bbe2804 : SUCCESS in 7m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/50def1333508446c931dd4c92acb1e82 : SUCCESS in 6m 40s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a519531c66f34a3287e51386a363c486 : SUCCESS in 4m 41s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/cef9e57d6fd94110be89a7343f441499 : SUCCESS in 5m 16s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5de88d6fd45e4a08a60c2756b6e6ee7a : SUCCESS in 3m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d5703d400a93464d95f93fec4d5e2055 : SUCCESS in 10m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7ec8ef9d61eb4fe78e0f6d5a59d81000 : SUCCESS in 5m 12s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/bd50bb6734254fd996917e50ea2e7e78 : SUCCESS in 6m 07s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9eaa97ae9d63469a98d4208db4cdcb7c : SUCCESS in 7m 45s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/f32b2975d79e470c97deaed1fb6d95c1 : SUCCESS in 17m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0aee380e4ea14a43a1495e31e40a6b4b : SUCCESS in 8m 35s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/376fc535ca6a467d86368498b0441f99 : SUCCESS in 8m 04s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dcaa22adb7f54e3fb61caa452bbea595 : SUCCESS in 12m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/21ff1d44db294f72a2dade82adf4f3ea : SUCCESS in 2m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/dda86fe9309147abb7baa676f0423aad : SUCCESS in 8m 01s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/15aac266784d42fa9674d29e5054bb0a : SUCCESS in 4m 41s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/aa4ca6916e9841378b17fdbc6c6adf59 : SUCCESS in 3m 31s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2b2f0f8ae5e7468e8c8b271d851e144e : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fca0523b7c57476f96326bff12ea5f48 : SUCCESS in 26m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6b9965b7bd884fa58a8ffb587726ac16 : SUCCESS in 6m 43s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5bf7b11298b64886b1101feddeda97d3 : SUCCESS in 13m 08s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/20a2990776174ac9ba026a6abeebda62 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edda516bd8c6428dbed17ba8e1bbbc68 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbdd7c0ee73f4ea5a21bfe85f012176b : SUCCESS in 24m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0d05103069704603a258e1fd020e530b : SUCCESS in 5m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6120f3cf0aa0448fa8e71fefb25fdc34 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee528a16c0af484fb711129d016f2a81 : SUCCESS in 8m 43s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/12633ed0a353418097632d75fe34a013 : SUCCESS in 17m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/34eb395b874c4b60b28c4bb4defee9e3 : SUCCESS in 5m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/590cf22ce25f4e81bbc86ceeb3e60b6c : SUCCESS in 1h 05m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5aaa52c825f44bf98bcf606cd69a5383 : SUCCESS in 1h 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"18e75ba5af9451f9ba4d4163e0d3edc9379302ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-11 10:02:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28b87fe59fc836abc4445991b09fbb6f1fe410c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 10:02:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c6ec2d0f2a7357bada71e6d53ad91ef5172c69d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:58:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4707eae459e5439cbcfb3967a92553c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56de1dd54de04a2eb515af358062ebb9 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c545b7c33f34d54a3ad3918e57a50b5 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28e4a687b2494da4ba3e383169b03438 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a1a4064fa4540d4934eb85e4f831f0d : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/969329c5e8044854a4e4d540cb00fdd3 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5a5eefc6b0f44a3aca03c9ef9e2f31b : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2aa8d2725f7e4633a79447a642b04da0 : SUCCESS in 4m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/646e82c5badd4e26b74925f5b4eb3b25 : SUCCESS in 6m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/59412149983b42fd806a211b8ee2fd2a : SUCCESS in 3m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5083b3351fb945c38b6b16c22fec2de5 : SUCCESS in 5m 40s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3068d7bc33694f56b0cda586c0d1ee90 : SUCCESS in 10m 33s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/96cb64f3549a414ea1c2160f0b07d860 : SUCCESS in 4m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b5241a89b8824f788684394803536caf : SUCCESS in 4m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/16bdb7ae17b441709dd0456060efcac1 : SUCCESS in 5m 16s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d15e957877da4402819ee549bd6b9c76 : SUCCESS in 13m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f9c1433cd4140418ab0e2c60117560e : SUCCESS in 3m 48s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bb6e46e3117f4405a980c5f9dea8a7da : SUCCESS in 8m 41s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4e69ce56515641e18a2a61095c25be95 : SUCCESS in 10m 45s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c77e20e0b65e414fbfabdde27e0c1648 : SUCCESS in 1m 55s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5b6e401b73c140378b4bb818095c5ef8 : SUCCESS in 8m 57s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0daf152b749540449d27bd43ab80440f : SUCCESS in 26m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e165b0ed2f346e0ac92a6652d546d5c : SUCCESS in 6m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/403786171e924b1d84921311bb7413a1 : SUCCESS in 8m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/664709d4d3064e5487dee91f4f85f5f0 : SUCCESS in 3m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/87a51fe927a44a099d29c2a2b1849286 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29b471a93dde41f19fe627fc67231d84 : SUCCESS in 26m 41s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/245815de03344c5ba5be1382433cf0e7 : SUCCESS in 5m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/951cb5e760b3461fa334d7c70ad9027f : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5d3ffe692cc4e378cbb6f89369c54d9 : SUCCESS in 6m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/745b9a75c7844131920e6eafe8ecac48 : SUCCESS in 14m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/937fc6b1516c42ca8ff32b4d371afa6f : SUCCESS in 2m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d89cc1ca2644806bd5993a25891ee48 : SUCCESS in 44m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b851c8e357f041499796db67984e1cb8 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6a50a181059c2d593113e48bf7954ee99956c364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:58:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"585eaa73905aa9657f128b45188c567ee1d40e7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:59:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe2f1db65e74b9596b4c9f4f06c9e43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ba3a66609d48dabcdc8bafb95234ff : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4d91d2ccde6209f531478db9ec6834ac056be04","revisions":{"e4d91d2ccde6209f531478db9ec6834ac056be04":{"kind":"REWORK","_number":1,"created":"2026-04-10 18:18:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/984106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/984106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/984106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/984106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/984106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/984106/1"}}},"commit":{"parents":[{"commit":"271d1a7861f04143f6b0eafedf885c9593cfa137","subject":"Merge \"Bump cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271d1a7861f04143f6b0eafedf885c9593cfa137"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-10 18:08:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-10 18:08:24.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 8.4.0","message":"update constraint for openstack-release-test to new release 8.4.0\n\nmeta: version: 8.4.0\nmeta: team: Release Management\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib4839c30b8d4c10b3ae6639ae14f947662a560d5\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9f5f9107f79422d77f246eb45ff81444ea3a1de1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4d91d2ccde6209f531478db9ec6834ac056be04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4d91d2ccde6209f531478db9ec6834ac056be04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
