)]}'
{"id":"openstack%2Frequirements~984143","triplet_id":"openstack%2Frequirements~master~I7f996b298b70a6a7a50d7b4db9058f8e9abd4211","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7f996b298b70a6a7a50d7b4db9058f8e9abd4211","subject":"update constraint for openstack-doc-tools to new release 4.0.2","status":"MERGED","created":"2026-04-11 08:17:31.000000000","updated":"2026-04-22 13:35:47.000000000","submitted":"2026-04-11 11:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984143-new-release","meta_rev_id":"24a51e714808d76e5dd071157a2ee79f39291faa","_number":984143,"virtual_id_number":984143,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 11:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-11 10:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-11 10:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-11 10:02:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-11 10:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 13:35:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"9c2b2dbd1d46b6fdb844e4ec4e7605bf124c4b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-11 08:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7ccb9eeaad04e5f1f14983d35ac11dee3521de9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-11 10:02:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62f486929b1ddd4e3201db05ef8ebd3d0d9f4f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 10:14:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2dc766afe474f4d8098b25ca831b4f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed5999d121641798e97116126d425fa : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844cf9fa78d342869b46b572b5990e90 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61fd8fb37f08443bac6fdcf355124fdb : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5d2728c75494f448f2bd69e6b19c815 : SUCCESS in 6m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5104b4564f664dda889ef6d4e91a4dfa : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5828acade6b14f77b0789033448cc4a0 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cec2e5a156704bfeab2d0719ec195fc5 : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c9d588fdc05345849f0ad2abb5436d0f : SUCCESS in 4m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d9f2b9066df746cb920e5c5530d1d2db : SUCCESS in 5m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7aedf7c6e5dd4eca962ec8d0f3d0d3ea : SUCCESS in 6m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7aa5d706bb8649fabc1b32bf0ed87603 : SUCCESS in 5m 51s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/db55e8dbf6114e38a5d7a812f520f6f7 : SUCCESS in 5m 31s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8509fa4a4f034f79b265e52a5c3968fe : SUCCESS in 6m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/288894e90bb54d17b48611eadc73f6c6 : SUCCESS in 4m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a05a243a753d474c8716245d2f7752c9 : SUCCESS in 11m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d0677aa8239345a19686f0f2e0e8d915 : SUCCESS in 5m 11s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1a62b35631cf4eeb83b07a35d7683ce9 : SUCCESS in 6m 20s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d0c133064b874969b5e7981b89496f0b : SUCCESS in 6m 53s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3f708bd9ddec489c8069a3ba9b4d2cae : SUCCESS in 15m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce082422bc514aacad7842577ee2284d : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0382ec9d08244bcba949a0b2ccdbaf07 : SUCCESS in 8m 28s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c4e96af94a4c4d628ba38c69786c159a : SUCCESS in 14m 16s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8013985dee184b709165d187500c01f2 : SUCCESS in 3m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/26d2c1b4564647e39693ef86beb65a55 : SUCCESS in 8m 13s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4b292749bae444df95d5f30b9dcbdc54 : SUCCESS in 4m 27s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/6a8b56dfcd2844a8954fd149f998b05a : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0d0e13c76c3746af84c58e6a10e129d3 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/bf729b4597e94609a320edaaf00ee3d1 : SUCCESS in 26m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8989b081f4e1449eb230eeb732c17ec4 : SUCCESS in 7m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/be12170ac42a4bea81eb4a136d056f73 : SUCCESS in 15m 58s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7aa20e0a9d2f42a788b8b606740ee4d6 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6720e2066da456cb684577ae3f5c22e : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04338eb548334e4cb2ca4ef21f878a12 : SUCCESS in 25m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/edfe5be0e91044c4857785426bcf35d8 : SUCCESS in 4m 31s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3085ef36ea48492c8b8f21766a8983fa : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b68120d22000443db04930660097ddd5 : SUCCESS in 7m 13s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f6c888d402434ca599d952f207c2a407 : SUCCESS in 19m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c257b93acfab440fab205329b14260ca : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87a018ee31074677a0cfc6ae3d21e360 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/153f459eddee426798b6bec88ede5daf : SUCCESS in 1h 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cc63e8c09abcfb04bf269636b911f695192e263e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 10:14:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7807b7e68720632cf4930a4c3ac524de827f95cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:58:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae1d4c707314218984e138b98900902\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76612f5e8e244d998b7c0f13f8ccbe09 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d5b4042b084324bf19730c98c103f6 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6db31af7d0b40a099af4160f2e2551d : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c0e2b27f38f4b1eaadcfecdaca432b5 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e2801b6e7584604961e6b745a582844 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/428198747b00422f8c5b3456023f8475 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0ff8050b955246908a28074f2ead37c1 : SUCCESS in 5m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/29188e73a98147bb8ceb219d10754e10 : SUCCESS in 3m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/af41717611074679aaec4523950db290 : SUCCESS in 3m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dbdefabc4d854417a3b6185bab8370b2 : SUCCESS in 5m 28s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/562307c6c4274779b40b0ee368c61cf9 : SUCCESS in 7m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/41f6ebedd82a4c4b9c6ba4601cf18fba : SUCCESS in 6m 27s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6d6efae5fdd34dbca36789e014a17863 : SUCCESS in 5m 03s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f618afa3b9c74e2ead38acc4f568fe2b : SUCCESS in 5m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b34a207335be43ca9b90428e152e0daf : SUCCESS in 14m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/117f1823bb7040b8b40d24ac76fec8de : SUCCESS in 6m 36s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e5e4a7fe8783472eb67a18ae7d0009c9 : SUCCESS in 8m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/da70e372b0fc4b6dbb6fbeb21d8c6eac : SUCCESS in 11m 58s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2eeb813d80544ca89429af71780d0c68 : SUCCESS in 3m 31s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7f0b1b5583424b5cb5939ce733ccee90 : SUCCESS in 7m 51s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4275a10f2aed4b56ab06031a2f4a2884 : SUCCESS in 23m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/235aa1ecf2fd4ff39312e7a6c4405229 : SUCCESS in 4m 02s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cd790b8f774946ba9ac96016916c4e8e : SUCCESS in 10m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/13fd2e7a65e848a89aa48d2ec6f4df03 : SUCCESS in 3m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/26d7b0f9166244cfa6c4fd794339154f : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94b6b72dc96e43c4a1be6ab83f9894d3 : SUCCESS in 25m 39s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ff3fa751704342f7b3a9df0acae3a856 : SUCCESS in 2m 44s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8f06df0ec006412296991b456a5fa041 : SUCCESS in 2m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d6b9ff0185342c1bfa2bfedaccd4167 : SUCCESS in 3m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ac7e4522c7a34254bb5d2ff293728877 : SUCCESS in 17m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3e7ddc18ff1746c5bd95b982aa052bab : SUCCESS in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb828a5b5a434f13993ddf149a1b9229 : SUCCESS in 33m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17971a312a034cea81abe4c5e34364c8 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3d66ac2f6981b2998b58eb39c0a700c039c3eb38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6b5d744f2cbb9fe0e822a27a5c57ce5686f9527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 12:00:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2241de8e494f3ca39031160667f439\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2012dd494029438280a0dd80c6b85211 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"24a51e714808d76e5dd071157a2ee79f39291faa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-04-22 13:35:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99cec3776c75a71fab595fbac1fde051c2fca7be","revisions":{"99cec3776c75a71fab595fbac1fde051c2fca7be":{"kind":"REWORK","_number":1,"created":"2026-04-11 08:17:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/984143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/984143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/984143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/984143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/984143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/984143/1"}}},"commit":{"parents":[{"commit":"271d1a7861f04143f6b0eafedf885c9593cfa137","subject":"Merge \"Bump cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271d1a7861f04143f6b0eafedf885c9593cfa137"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-11 08:07:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-11 08:07:30.000000000","tz":0},"subject":"update constraint for openstack-doc-tools to new release 4.0.2","message":"update constraint for openstack-doc-tools to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: I9a30f735a1e84388402ab9e1f6696a84c3c2cb01\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I7f996b298b70a6a7a50d7b4db9058f8e9abd4211\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99cec3776c75a71fab595fbac1fde051c2fca7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99cec3776c75a71fab595fbac1fde051c2fca7be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
