)]}'
{"id":"openstack%2Frequirements~984152","triplet_id":"openstack%2Frequirements~master~I59857d930e4ac04a90d2e8e10c360b4f30fa1a13","project":"openstack/requirements","branch":"master","topic":"drop-pysnmp-lextudio","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-13 07:31:33.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I59857d930e4ac04a90d2e8e10c360b4f30fa1a13","subject":"Drop pysnmp-lextudio from global-requirements","status":"MERGED","created":"2026-04-11 12:45:47.000000000","updated":"2026-04-13 08:46:31.000000000","submitted":"2026-04-13 08:44:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984152-drop-pysnmp-lextudio","meta_rev_id":"4d943a37eff7085c5a2ac9ee0cb8e06c20c01fe2","_number":984152,"virtual_id_number":984152,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 08:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-13 07:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-12 18:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-13 07:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-11 14:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 17:20:26.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2026-04-12 18:47:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a590f830e7d7d35b27743a23a14e0d2388b8c488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-11 12:45:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d03c3042c711a72c7a9bb7bf92f9c59fdc2ac568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 14:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529bb58b20c24eecb01b19a1d2522a9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9633b94a81f942c383a3aaaac9138fec : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f04c107a4b4a1e98c6ce0611755d0c : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54ee1fa4c2d244e480e5b92350a1528c : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45d464101e3f4152b16c6db947f5b2f3 : SUCCESS in 7m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04c5dbaf13ef44c3b98ad3c7d924e6bb : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4956dc3941b48e8a0dbf59f26b4841e : SUCCESS in 2m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d8a035092da4d3a9e7cbebe631308a1 : SUCCESS in 2m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fbbcd30130c44e148c00837b170627cd : SUCCESS in 3m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3ff9b640755b4904b5325a05dddd6449 : SUCCESS in 3m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5436425096cd4f44833b457e80df302e : SUCCESS in 4m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad10fbacfcc04321a3332525c03a7769 : SUCCESS in 5m 16s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/cdba5503451849c2982731e09210246c : SUCCESS in 2m 38s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/bccae6b73c914784a5b64fecf195c9bd : SUCCESS in 4m 27s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/7e5810d4aeb447aaa68dea3f75674824 : SUCCESS in 3m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c72003e417c749249ad6d2034a43cf56 : SUCCESS in 8m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8ee88ef96c2b4acfaa50a129afaf0f15 : SUCCESS in 4m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/dd5526840c964e0ea5200cb033c79644 : SUCCESS in 4m 55s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ae879ead35fc4279b1853a36e8979e2f : SUCCESS in 5m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/95e369e9fc24481caca0dc8e397d5b8b : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9bb96033f0944f98637a43b828f3944 : SUCCESS in 6m 36s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9c3794ae1a264198b7215e3104987480 : SUCCESS in 5m 54s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/bc73b8021817424eb4c02e42d98429fd : SUCCESS in 10m 56s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9ca3a22080fc4eadba67b326fd70fcbb : SUCCESS in 2m 04s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/291303a9f31c4bc99b2b6d4b090dd9ac : SUCCESS in 6m 37s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/922f0735a4f14b8ba4a8bc22730db36f : SUCCESS in 3m 30s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/caee7bc8b485414ea9fd9e2b069ba21f : SUCCESS in 4m 48s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a10d9e609b384ee9afa704a12a2222bd : SUCCESS in 9m 36s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ce89ca5d26ba451f9a7235b5bf1c293d : SUCCESS in 25m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b9622613cd7f4168b706e058b2d88b93 : SUCCESS in 9m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c8ef312241ed483cbdc8343f55c48761 : SUCCESS in 11m 12s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0f40f015a0444e98b69b5dd0f1e86967 : SUCCESS in 2m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/111e2360d15443dc9587c68f87b6ebbc : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf54e41fa9bd493bacf03c7e2de0a731 : SUCCESS in 26m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/410b29569ddb4676929613d7651b9743 : SUCCESS in 5m 50s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/87b8e8b62196460eb36cdc0bc779049f : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85e390ac7bc74c229e20d3d5216f15fb : SUCCESS in 3m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/cfae3d3087604a7e9decf3677e91964d : SUCCESS in 8m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/246508ed439d44cb9a0e7ab001c58bfb : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62a881c1b12247a8ab734e8718c2064f : SUCCESS in 1h 08m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86032c811a9b40aaa7121cc54814e7a9 : SUCCESS in 1h 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"81c650caa4ec83d8ec00083db2a7feac50270a63","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-04-11 17:20:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"604b80d4e7d4a06fef8a440d99bf20aaef18513b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-12 18:47:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3af398a2fba8b4a6a34cc045bf412cef7c1abb10","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-13 07:31:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44a001af76d2fa80408cbe23c8ee7eb79a2252df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 07:32:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da3c44f5b544a512fbe9871695b592ce34d38b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 08:44:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219fe3df3410458d9b88223ccfbacffb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a14bd8368f845e49a56c4df7c965c6e : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58073314fa174316856fd5d9bcccd15f : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf939cfe40804c10a2e41bf7c3c77a06 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b86ac416b3434ecb86c77fcda4b83f52 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/827d121ccb3842779e63880d4ae8de54 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e3d613031b94bb89896da2e25ceec2e : SUCCESS in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/08d3ee544ac44a2ead72116a169801a3 : SUCCESS in 4m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d1802f07a9a74d178fe925e032a4eb7f : SUCCESS in 5m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a9ba2d126bb24de0922a769e8e5c3699 : SUCCESS in 4m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/565931c9035f41dc8d192e0dea3b439d : SUCCESS in 5m 46s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/1aeb3936bbc040a0af51ae84359612c5 : SUCCESS in 11m 25s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8c5ed9770bad4a01b8a23a01b579aa32 : SUCCESS in 4m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d6f9a81e3b34433c811ed2508223dbfd : SUCCESS in 4m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8edea5f791e24527adcd87757a2e430e : SUCCESS in 6m 15s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/aa9ece0d38a74d6c81ce29b1e5b8c8d0 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eba3a31a6c664bbabdab8d12cba012b5 : SUCCESS in 5m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/fc3bf84382b24fa6817acc308ecf32e9 : SUCCESS in 8m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/67e02b5290b7446d8bbca2f54fdc5549 : SUCCESS in 9m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f55e6061ce164378bb09814e70a68047 : SUCCESS in 2m 26s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a9ef85bb4c104be48c96b718e8e8926f : SUCCESS in 8m 10s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c47f9e1763184ac19d4e974cedc0f8d9 : SUCCESS in 23m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/14eb77bba4da4696b9695ac2681f4d5f : SUCCESS in 6m 12s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2b7b5610cbfd405c8524c6517b1cae3c : SUCCESS in 14m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/28255197fde742a28d33498098a501ce : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3a9336b3c5e4754a87b4b2ea91922fb : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71b0ed76162f4970a4c045d9c4d71f1d : SUCCESS in 25m 08s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/eb13ad8e510943839e9057a31d79aa9a : SUCCESS in 5m 52s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e9ca30d2383843b1a24360efe9f66fed : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0d79ab25979486ebc1c0aa3ebcdc89d : SUCCESS in 6m 42s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/42c7255b09734ad19353d833f21bcfe3 : SUCCESS in 10m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2531faffc8544bd6a38b87af43a71a55 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad7cfeba770448f4a7e4bfcfe0dfb98c : SUCCESS in 1h 06m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7717e7143f394efbbed4e2f08b200157 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"72ba613dd4ce7f6a59dffacdc2f41117b3e04d14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 08:44:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d943a37eff7085c5a2ac9ee0cb8e06c20c01fe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 08:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d8b0016d7f424d96ecfadd04a0efa0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118f809233db476cbb6a44852166d94e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7acf8387cbc37bb228b2a075de6c5479edb5f290","revisions":{"7acf8387cbc37bb228b2a075de6c5479edb5f290":{"kind":"REWORK","_number":1,"created":"2026-04-11 12:45:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/52/984152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/984152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/984152/1"}}},"commit":{"parents":[{"commit":"45df797841cb4e20bcc0fa3f31df74977321933f","subject":"Merge \"update constraint for openstack-doc-tools to new release 4.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45df797841cb4e20bcc0fa3f31df74977321933f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-11 12:43:56.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-11 12:43:56.000000000","tz":120},"subject":"Drop pysnmp-lextudio from global-requirements","message":"Drop pysnmp-lextudio from global-requirements\n\nThe dependency has been dropped from ironic in [0]. Also the comment\nabout pysnmp is no longer true, it is being maintained again and vitrage\nis using it.\n\n[0] https://review.opendev.org/c/openstack/ironic/+/982147\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I59857d930e4ac04a90d2e8e10c360b4f30fa1a13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7acf8387cbc37bb228b2a075de6c5479edb5f290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7acf8387cbc37bb228b2a075de6c5479edb5f290"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
