)]}'
{"id":"openstack%2Frequirements~984222","triplet_id":"openstack%2Frequirements~stable%2F2025.2~I2e49dfff6bbc3d7de0681563cd7d93edbc48b167","project":"openstack/requirements","branch":"stable/2025.2","topic":"clear-legacy-jobs","hashtags":[],"change_id":"I2e49dfff6bbc3d7de0681563cd7d93edbc48b167","subject":"stable-only: drop bionic wheel cache job","status":"MERGED","created":"2026-04-12 10:37:16.000000000","updated":"2026-04-12 11:42:49.000000000","submitted":"2026-04-12 11:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984222-clear-legacy-jobs","meta_rev_id":"2e48da24732fd25de0ed2b45633a8c45bc637761","_number":984222,"virtual_id_number":984222,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-12 11:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-12 10:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-12 10:38:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-12 10:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-12 11:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"953723723420707098b6eba94afb2128394ceeec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-12 10:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"498fa158cb6f1579ee754629ab0bc660a2576619","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-12 10:38:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a30158a283e56c0b0f2defa55f56c85088d15058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 10:39:54.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c09862ae8b58d05dd03716794073f7e6d3ff3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 11:42:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a25ce62c0942be9f22edff1d5a3649\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cbb15e2d1249ac893b7d4adcb1a9f7 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/083656da886742bc9ccbc0a9803d92f4 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43cb35ce59834734b7027c0379660094 : SUCCESS in 2m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46840b75e89d4343979def2c95eedd2b : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/674bd6d3632345cf9e42b9764138662e : SUCCESS in 58m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1d69bd8a156439f81b323dad0904320 : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2e48da24732fd25de0ed2b45633a8c45bc637761","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 11:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7690d7e7bb3671a425b2332d92977deebf3eb00","revisions":{"e7690d7e7bb3671a425b2332d92977deebf3eb00":{"kind":"REWORK","_number":1,"created":"2026-04-12 10:37:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/22/984222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/984222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/984222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/984222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/984222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/984222/1"}}},"commit":{"parents":[{"commit":"56b6bc82479c2a2e5a4b72e2cde04aa20673853c","subject":"Merge \"update constraint for horizon to new release 25.5.2\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56b6bc82479c2a2e5a4b72e2cde04aa20673853c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-12 10:36:00.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-12 10:36:00.000000000","tz":120},"subject":"stable-only: drop bionic wheel cache job","message":"stable-only: drop bionic wheel cache job\n\nThe ubuntu-bionic nodes are no longer available in the CI, so this job\nhas been deleted. Clean up the dangling references to avoid zuul config\nerrors.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I2e49dfff6bbc3d7de0681563cd7d93edbc48b167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7690d7e7bb3671a425b2332d92977deebf3eb00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7690d7e7bb3671a425b2332d92977deebf3eb00"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
