)]}'
{"id":"openstack%2Frequirements~984233","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I2e49dfff6bbc3d7de0681563cd7d93edbc48b167","project":"openstack/requirements","branch":"stable/2024.2","topic":"clear-legacy-jobs-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-22 21:09:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e49dfff6bbc3d7de0681563cd7d93edbc48b167","subject":"stable-only: drop bionic wheel cache job","status":"MERGED","created":"2026-04-12 14:13:53.000000000","updated":"2026-04-22 21:09:34.000000000","submitted":"2026-04-22 21:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984233-clear-legacy-jobs-stable/2025.1-stable/2024.2","cherry_pick_of_change":984232,"cherry_pick_of_patch_set":1,"meta_rev_id":"c15e241a6f0ff00a88013d17000e56a68d62aae5","_number":984233,"virtual_id_number":984233,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 21:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-22 19:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-04-22 19:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-12 16:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 19:14:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"264c233478661e59c3ef5ec960f33c5558f73152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-12 14:13:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"348f9c9da3c318f1ed27df38fc8d5d6584bab18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 16:04:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18bd7b58c5c944a8b52752f0dae6a71a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d456fc7659dc4734b47a6c1c0c5b5cd0 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a2cb8c6a378492095f40251c299bf60 : FAILURE in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7c6f1a16a244cfeac5419977780de67 : FAILURE in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdc57e8640dd47fca5bcf75f2a6ed49c : FAILURE in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b176a2b0f7e4039954b791d320b0e80 : FAILURE in 2m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ef7f56d2045415e8565c6f97c8ca4d7 : SUCCESS in 59m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31d0a52636184a978588dbdf04a5b5d8 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"992a993e17bd5c212711ab70d125b3f61f3e4787","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 15:27:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f6cd912548ca4012a3d5f11a56d07e0e32df639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:39:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/427094e6b3ee4d3997e144f583309a74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c98f7e687a548488a74cbc2f9f9375d : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/906593032e30475899db92cb464ba3ce : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62f441a7375d4a02ab90db89c7b6b89f : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e1cccab09d24cf28b7ee3c4ba28a214 : FAILURE in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a856b3acc63c4c57b25a63e73e05514d : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b18229b1b52340878ff3712ea94fc60f : SUCCESS in 1h 02m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67784b1c381f4a9996bc6a327a09cb26 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":1},{"id":"769382241945a0da164af4f4baf406f305830836","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-22 19:14:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27d9bff7218dea7fade2576337b3ac2ac7c94a4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 19:16:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59a7444be77e44b9fa2722018a902b62f52967c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 21:09:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddbe7f7d88a4f22b8d15ebe703bf770\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660eb0858ba34f868ec115ee5bf8f7bd : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57924b54c4a14021892acd93e8ac7c1f : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b636c9a3a48348a38867d01bf3631446 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/051d8fa8c6044de8b6282b05d22c4033 : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79a63416246647a787f08dba4e9d7f5b : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55924895eb2649a9a8b95430b979f188 : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c15e241a6f0ff00a88013d17000e56a68d62aae5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 21:09:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7b93e87ba1ca4a152e36e9c8c93ac256a7aa4bc","revisions":{"e7b93e87ba1ca4a152e36e9c8c93ac256a7aa4bc":{"kind":"REWORK","_number":1,"created":"2026-04-12 14:13:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/33/984233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/984233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/984233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/984233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/984233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/984233/1"}}},"commit":{"parents":[{"commit":"92271810d9e66e3f8d5316c88b7d7bc95d293351","subject":"update constraint for os_vif to new release 3.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-12 10:36:00.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-12 14:13:53.000000000","tz":0},"subject":"stable-only: drop bionic wheel cache job","message":"stable-only: drop bionic wheel cache job\n\nThe ubuntu-bionic nodes are no longer available in the CI, so this job\nhas been deleted. Clean up the dangling references to avoid zuul config\nerrors.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I2e49dfff6bbc3d7de0681563cd7d93edbc48b167\n(cherry picked from commit e7690d7e7bb3671a425b2332d92977deebf3eb00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b93e87ba1ca4a152e36e9c8c93ac256a7aa4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b93e87ba1ca4a152e36e9c8c93ac256a7aa4bc"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
