)]}'
{"id":"openstack%2Frequirements~984374","triplet_id":"openstack%2Frequirements~master~Idd14bef186116f7691f9e64237885735aead2c06","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idd14bef186116f7691f9e64237885735aead2c06","subject":"update constraint for openstacksdk to new release 4.11.0","status":"MERGED","created":"2026-04-13 12:01:04.000000000","updated":"2026-04-13 15:20:02.000000000","submitted":"2026-04-13 15:18:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984374-new-release","meta_rev_id":"898bdacee3594aea0b7ecd120f0a21e907fb0ed3","_number":984374,"virtual_id_number":984374,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 15:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-13 12:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-13 12:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 12:42:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-13 14:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12a396880665b7836c8f3f2439bf4deab52a7f6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-13 12:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c4f5257baa286d80786f1cc02388a973d67a0f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-13 12:42:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f3b03a98b77d2fe12a109ccb742d0d20963b074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c42039ac7e045c987e095fcdf5b4d6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4593931d43cf4c798e82749316f2dccf : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e9f924246a4be6ba52896e8afc93e7 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4169eeb8c18f4c7286970a848f453542 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d6dac6321f747dc95027485830be247 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf366cf6a59c433aa3ca35a61ce7b6d4 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3eac969604bb4a39956a6d7f05ca3185 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb185ba983dc4c05ba42be337c6ce140 : SUCCESS in 3m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f6bafa639c2949259b322f62f8d344c1 : SUCCESS in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/353fa0184a06491a93962581d9f0237b : SUCCESS in 5m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/da08daa8d2d54df8bc38138c857e1c49 : SUCCESS in 9m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/61b1ad8f12f94da5a0877d968c074bbc : SUCCESS in 11m 05s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7d423b1d0fc44ce4ba1917ff5657c338 : SUCCESS in 5m 17s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/277dd8d24faf46ccbf2d53ab240c0f50 : SUCCESS in 4m 55s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ef729d1e8ca347f492c6c11d32f409ab : SUCCESS in 5m 02s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/7b29766cf70b4f05ac9f16e958f403a6 : SUCCESS in 10m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c1da54932262402fbb3841a1075f0fd6 : SUCCESS in 3m 58s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a80666d3b8404ab391b714e5110afd03 : SUCCESS in 4m 55s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b5500a4b4dda4c5fb379cf514f45b5a9 : SUCCESS in 5m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/fab4301a846543e38b7894d958dc1eb6 : SUCCESS in 15m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddf210da53674ff98642c7d32b1521f0 : SUCCESS in 9m 13s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/17f9d218827f44748ee0f8753af76a60 : SUCCESS in 9m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/552dfe5ed0f44b1aac68c5a757ef8fa2 : SUCCESS in 12m 05s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5baab659410e4367b39ba19225936c83 : SUCCESS in 3m 43s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/de0ca5c4644d47568575c918cb7683bc : SUCCESS in 7m 40s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/50a0e225a379451a830c270a67973f3b : SUCCESS in 4m 12s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/b72f13a4091b45dc9b76648a47b0fdf5 : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4183487ad7194273ad85d2303ba4dd93 : SUCCESS in 10m 18s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d1c3fe9faaa4454f824efa030654c20c : SUCCESS in 25m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dfe13cecea9e4e07902efc644c5323db : SUCCESS in 5m 58s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/263e5b8e03024bf785c98d595223fe84 : SUCCESS in 14m 52s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/aabd8e4e7bb04892a907afa26ad48b2c : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc07341dea7d4ff78f2a720d0f42f957 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab15a7c0b58c42298f6c61a20ef2e12e : SUCCESS in 24m 07s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/af417b2e05424f1f84bbb66a94cf2b6c : SUCCESS in 5m 55s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f2d5ab3257cf49848b0da114f201007c : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6aea96c11f2747378e85e45fed25cc5f : SUCCESS in 6m 25s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d527bf7827c64b539c5dc8b595c9384b : SUCCESS in 16m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6e8a46e880084e919b2af5c0be1694c8 : SUCCESS in 5m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4d267fd971f4e54836c9066c2a864a0 : SUCCESS in 1h 06m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b90330b1a8ed414bac7f2b70abb8f341 : SUCCESS in 1h 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"eecb8b764127595dc8000fd219e3a42dcaea4efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:05:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6fe37fbc93e289a1ee480d815168a49ef528ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:18:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c606266478644fcb30d04ac387a192e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b0dd3fd5344bf1b034a84ccba5283c : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec7e45e50c24a95a9b7b25a8407d06a : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f6b92826f354223aaddfbac8fc4c10f : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a6099fd2f314daa870c3daa61d7d0b9 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/857a2d0925a942f19667fa90517cd24d : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59ba00cd181342dc9a06a93719d6dc2a : SUCCESS in 4m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a06b24dfb5fa45ef8699453363ea0154 : SUCCESS in 5m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/189d132da1fb4d86afae5284ea2ff3df : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1c363df8b4114ae0bf5eb8914bc0f7dc : SUCCESS in 6m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4a6f0473a5544a3db11372574b38048a : SUCCESS in 3m 18s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0da63f71d914431b8f2be36d419886a0 : SUCCESS in 7m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/083587bc859044efa232a5d056b4678f : SUCCESS in 3m 08s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3cfaafaf30d840acb79d8f5f9326b15b : SUCCESS in 5m 08s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/73012862d4b54036a80ccc03fed7ffc4 : SUCCESS in 6m 31s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3caca25a2daa48c7b809fb535f27ac32 : SUCCESS in 15m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67cf0f969da54057a7cabf72516c462d : SUCCESS in 6m 51s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4aa86c4d222d4d6182c343da5cb5ae5c : SUCCESS in 7m 47s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0c63ca2f9a23476cb076cb24c24f6ea1 : SUCCESS in 11m 03s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a07f796c99a9437d94fa027aca251e9d : SUCCESS in 2m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/841e7773dc4f4fedaef133a395b7e0ee : SUCCESS in 8m 11s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6c327924a087447795a3558877e87bed : SUCCESS in 26m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d7da823dbb534b21a8ef013bbd5c348f : SUCCESS in 8m 04s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7942fc6ed669445ab4e37688ca146e3e : SUCCESS in 11m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d3843d7532f0465f811b6b2f0b8166ee : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db42f88420d54b76b98a7036704ea518 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c411d454e3554cf594c5122d738d5523 : SUCCESS in 27m 30s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/cbf472f6dff04260a6fed08b989cfba5 : SUCCESS in 5m 49s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/011b509f91504abdbb55ce4bf403c2cc : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f3af49b304b4be9b940e6b1b8ad5831 : SUCCESS in 6m 47s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/dbfb271e41b14ec7a50afa13ce94385f : SUCCESS in 18m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/51e65769237e4495999f3df25b010351 : SUCCESS in 6m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69fcbe802bd64596a7bd10369d591b6a : SUCCESS in 1h 08m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ce4888e8e6b479da1862a45c11b5683 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2782f057dac6d374ee74e03522a03ba3c45df10f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:18:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"898bdacee3594aea0b7ecd120f0a21e907fb0ed3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:20:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56de6bceb8d4b0aa1012042c3b75ebc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4049eb88ad408ca03ac17b78850bce : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84e467a02fc484ca063ae383ab95d04f530090fb","revisions":{"84e467a02fc484ca063ae383ab95d04f530090fb":{"kind":"REWORK","_number":1,"created":"2026-04-13 12:01:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/984374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/984374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/984374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/984374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/984374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/984374/1"}}},"commit":{"parents":[{"commit":"2c6eae3260867a648c252e8f39520cef685aba65","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c6eae3260867a648c252e8f39520cef685aba65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-13 11:51:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-13 11:51:01.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.11.0","message":"update constraint for openstacksdk to new release 4.11.0\n\nmeta: version: 4.11.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:e8b69cf27 Identity: Support authorization_ttl property for identity providers\nmeta: release:Change-Id: Iaf84a3a919d1d01b351ee6e01ad62205f39b7e6b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Idd14bef186116f7691f9e64237885735aead2c06\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
