)]}'
{"id":"openstack%2Frequirements~984381","triplet_id":"openstack%2Frequirements~stable%2F2026.1~I8e091e70829078b551789567cfb5f5570dc589e4","project":"openstack/requirements","branch":"stable/2026.1","topic":"new-release","hashtags":[],"change_id":"I8e091e70829078b551789567cfb5f5570dc589e4","subject":"update constraint for oslo.concurrency to new release 7.4.1","status":"MERGED","created":"2026-04-13 12:51:57.000000000","updated":"2026-04-13 17:59:41.000000000","submitted":"2026-04-13 17:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984381-new-release","meta_rev_id":"b8c53aeff3100d3382cf34197adab092a1ae56ff","_number":984381,"virtual_id_number":984381,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 17:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-13 16:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-13 16:02:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 14:47:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 16:02:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a8e34cfe41675bec3718d46128269a960583d8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-13 12:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1afd3a3e43431a9f59b6e0eca3e93b4429026c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:47:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a182063533624f4083007aae99c752fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5147705cfffc4d5bb0ff05c0aee30d14 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fce331a268c4807ae65709e04b6093e : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3812d16141fa47c9897fea14dd4bc53b : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd448e8d3c3f468388ee266aa47d99b6 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da833a6be6a7409e8c37b7379c77033b : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/09383c8093ee4614ae0f641131f4cfdc : SUCCESS in 4m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/aa7faee8120240b4a1241a149f72a7dc : SUCCESS in 8m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/894fbfaef3de4a068246a6aba1071661 : SUCCESS in 8m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/270df82e22684b85b8326b52d672a53e : SUCCESS in 10m 24s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d101fd9074044301ac34d47d33a0fe96 : SUCCESS in 5m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/263a1d9cc15143ccba8f0bb9f5a7f950 : SUCCESS in 6m 39s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2b14564072054ffc8f5e495f818af3c1 : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6bda97f9882c4cab8f94d9be6a8d6214 : SUCCESS in 13m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/605419a691404533ba04aa223edc123a : SUCCESS in 6m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aabb160c34aa428b9c75cd72da05f0cd : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7696d21717d44079aa8f0a0e28ceed4f : SUCCESS in 8m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2cca90a49a344ceb866b365760b9d4f3 : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40e759da140d4fc9a37e936f743e7250 : SUCCESS in 9m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/68e9c34689b74789b271f16193946efb : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b40245f3963345c58b0bbd37d6270d19 : SUCCESS in 12m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/04d27a5d7ebd41c2bcdac04b03fad6ed : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ff08eab231943a0ad2dd00623614494 : SUCCESS in 13m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/42fb2ef4b6864d3eb1e261b55e23183b : SUCCESS in 4m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5fe4aed64dfd486d8455ddd94cf96170 : SUCCESS in 5m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b5971b14ea504d0f965b3dc56f7f232d : SUCCESS in 11m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9b23a739ecb7462580c50afcf4574e13 : SUCCESS in 29m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f137831c1824421781c09e158a9ab29b : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4249a32c172c40549ecde3018b5a3b7a : SUCCESS in 15m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4e86de5565254c6ba7731d48bb097c8b : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba172f04f3fd455bade7f5a888de3769 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77ee2ea96f594e26b15a03a2ad8d576f : SUCCESS in 29m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/61d68545f0e24e63a67162534c745c27 : SUCCESS in 7m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/186a59c6e16f46cfb6c181520f0421ed : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9cd94f9f9b5944d290e4b3540ca5f0d5 : SUCCESS in 6m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/06b2a6790e36403ebef3d67b7ac659f1 : SUCCESS in 15m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a59d00449a874156b290c4787c8a0ee8 : SUCCESS in 7m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2c37c911cc54b1590780b2e911a09c1 : SUCCESS in 1h 07m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07c7bf0e537e4b13ad6c6f926b14aaa4 : SUCCESS in 1h 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c465fe4b0eb992719d901237bb1ed25a029c7f74","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-13 16:02:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbaf6a6cc9ab76e5829309438de6f95c497ade81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 16:04:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1ec2495138e80ca1406d77b186a1024d3121a4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 17:59:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0811383141c418c9227cb85339c05e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e18152217745b0800aa67d45a5b33b : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5cbf138ff934b8faf1ace06b0c7787c : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5d28aa800344ed9bf3e171a66cea7de : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/070f73c218894326ac834c3c4d609443 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3b75e8063184e9db960bb49f6399303 : SUCCESS in 3m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e56960313e0b4f4cbb4448e42dbbaa9b : SUCCESS in 6m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4f75b5fe89804e5cb30974fa00fc06d3 : SUCCESS in 3m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6216fefed7e64f40834ee5813a7a6dfa : SUCCESS in 6m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/69f4a10bf7944e9bb751f4f4df2635c6 : SUCCESS in 6m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3bfe05f973bd413188c0cdc9c97d1e1d : SUCCESS in 9m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/36a1c977106d489fa54016cd77654aaa : SUCCESS in 4m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7ddf57b824fe4ff68c138cd9fe07245d : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/afe36ec061cd411d95cecd70f49331b7 : SUCCESS in 7m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/799ab7215fe342acb5dfc3280b5ae8bb : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e002dcb6f63e48f895c446540945beb8 : SUCCESS in 9m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c0117bc8e0144a61a9770c6d45045950 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bba94701616b41baac13c5d0b2d01dcc : SUCCESS in 11m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1154048c672948ee8463e5ad17a8b1cc : SUCCESS in 2m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dd0f9a8474364c42ae6af0baea5a9e86 : SUCCESS in 11m 17s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8d339069d11f4e8e8bab007a076455c7 : SUCCESS in 18m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/809af12e83bc417c911af6a1cf3ba62f : SUCCESS in 4m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9246b57e5f3046a7882c32f32df86a3c : SUCCESS in 12m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/414fa389ff9b455f8c448bd1af6c3b78 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fed6c5d167af4ed49123183a175151a8 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2d573dc58374af383913ae5fc22e0f5 : SUCCESS in 27m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ce8f9672d15444449aec71cac46edcdc : SUCCESS in 5m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/43aff61632ba45eabecb3238d74d1675 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55cb3d009db545a4b1f2fb72e6d6cb2c : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/30cbce19dd464b84bb488777b79354af : SUCCESS in 19m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1e4666e961884463b005695c0add1605 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/848eebad7adf4dc6a9f8dd44322339c3 : SUCCESS in 1h 27m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6eae612ac98f47bebf3ce95263d9796b : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b8c53aeff3100d3382cf34197adab092a1ae56ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 17:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e07cb254ee7786723be7030ff3fd438e2960e2b","revisions":{"7e07cb254ee7786723be7030ff3fd438e2960e2b":{"kind":"REWORK","_number":1,"created":"2026-04-13 12:51:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/984381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/984381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/984381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/984381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/984381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/984381/1"}}},"commit":{"parents":[{"commit":"c9b3d86faf2ec93570bb843dba2277647dc5b06f","subject":"Merge \"update constraint for cliff to new release 4.13.3\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9b3d86faf2ec93570bb843dba2277647dc5b06f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-13 12:41:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-13 12:41:57.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.4.1","message":"update constraint for oslo.concurrency to new release 7.4.1\n\nmeta: version: 7.4.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ib4f005504e94e9b5ef220b3d1226d630ec919865\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8e091e70829078b551789567cfb5f5570dc589e4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e07cb254ee7786723be7030ff3fd438e2960e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e07cb254ee7786723be7030ff3fd438e2960e2b"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
