)]}'
{"id":"openstack%2Frequirements~984552","triplet_id":"openstack%2Frequirements~master~Ie4428c5ee444a51eac22dcd935172b2160b5b56c","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-15 06:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ie4428c5ee444a51eac22dcd935172b2160b5b56c","subject":"update constraint for osc-lib to new release 4.5.0","status":"MERGED","created":"2026-04-14 12:20:33.000000000","updated":"2026-04-15 15:44:56.000000000","submitted":"2026-04-15 15:44:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984552-new-release","meta_rev_id":"2e37628c7080b47d3af100089bcb8ea7bb2d73c4","_number":984552,"virtual_id_number":984552,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-15 15:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-15 06:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-15 06:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-14 14:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-14 18:43:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-14 21:07:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-04-15 06:59:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-15 12:01:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"146ecee5e27fca085bba89347287d4bbe173a5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-14 12:20:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8545dddde435b3f619c2ff5b138a18afb15c1e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:12:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bbfe6accd9c41ba83c1e8e5af749e4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769c42bdef1948edbf467fb9d40420ee : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1b4b503bbc46e8a31c04e57a1d436f : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eedecc7a977d4f32ae862d97f9f78d28 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17f5788f934f4203809b26243f40a71a : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/31423514bcc441bd9f5d3cf575e307ff : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/921263c23e0742bf9b4905be3afe6079 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8470a77fd62f450a80850c7b0a2bbfd1 : SUCCESS in 2m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/790c14c7db5f45f38df3c8086d43d2a4 : SUCCESS in 3m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98bce1a4328c4551bf7eece5c3a51b76 : SUCCESS in 7m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f4474c1b0f8440ca8dc28bc4fefd6657 : SUCCESS in 8m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/23880d0c73844f69929a8f03fa0976c7 : SUCCESS in 5m 43s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/8c0b83d2df9643fda479329aa8a87038 : SUCCESS in 4m 24s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/654906df12934027b1404b1374cc3f9c : SUCCESS in 4m 54s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/04667de8f343433b86f22962c2afb802 : SUCCESS in 5m 34s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/daa7f6779baa46748df8be40fddf798b : SUCCESS in 12m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d74d40394c364912b6d53ac7adbd6e55 : SUCCESS in 5m 01s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/36301fff25e64fdba899c094c1ee231f : SUCCESS in 7m 17s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7f070470e8dc4b87acea0f1cee50b2a1 : SUCCESS in 6m 29s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8bcb8a1d27bd43cd83fb8ea9eb4d85c7 : SUCCESS in 13m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cda7272067c64d7e95c882d63d813115 : SUCCESS in 6m 08s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/80311e32d20946d796a386baa92ab287 : SUCCESS in 7m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1888e343b71b4f069399bd2419b3199f : SUCCESS in 11m 43s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/3a53397439874debbd27e583598065df : SUCCESS in 2m 44s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a9a9b43628a948af80dd728a053b0811 : FAILURE in 8m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e1512b11995a459e85cdccc8d39fa63d : SUCCESS in 4m 06s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/fa00feeb0b5a45b9a5f478c8708979c2 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4606b118963842b1b7bed70c7198d6a6 : SUCCESS in 10m 18s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/93ceb63e2e914ee6aad7aff3af8caac3 : SUCCESS in 19m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4296267958a74be1ae604ef819b31371 : SUCCESS in 8m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cf21e42dad4b4644b5e7abef18e8bc7b : SUCCESS in 15m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/43d9c780321142e1b0732a22fac8908d : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8871c6e74ffc4c758cf687bfd119f9f6 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a66bcc8b68594d499abf6026cf11b0e0 : SUCCESS in 27m 22s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/62e166f850f94e98a2f2eed9e5022edf : SUCCESS in 5m 09s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/028e5c5036a6478f8981bebea5a34ba0 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb843f574b7a4410a24d1b51470b7c90 : SUCCESS in 6m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e52f88f20db3486580466b674b9b7da0 : SUCCESS in 13m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb41e7d330044802a77aa0b53d11cde0 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/018a6a27607c4bd58c72e354ddc53bc2 : SUCCESS in 1h 06m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76f5063d670d49b4a9a54253a41b5f12 : SUCCESS in 1h 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1e7e3f964e7aa3f064bf9347306da0ecb4b7e222","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-14 18:43:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bf4806d7e829dd6cd166cb3301a6ffbe3d5ebc7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-14 21:07:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00ace3cdbb5f324551e2a283750209f66611e6c9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-14 23:50:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfdcb019ed441bc8bb4b05e4ff4d30ecd0273bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 01:20:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66063e552ec845469779ba9849c29c8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4b4f230b614b0799964eeea551b3f6 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57311da149b3485188d7b55df1cad776 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77396ca3eb99420089e0b1b245ad8a85 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa226376ff644a649e050c13c579c5d6 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f46420bca2a446187f13795ab929ff5 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9befbad3b239448580621739b36e80cf : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0cd5763b24341e28dc5898b5a27d259 : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3021b4e4e3a846739310a82202545611 : SUCCESS in 5m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f91621cb26ce40cebaddd6c420534075 : SUCCESS in 7m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0b7719cc46bd42908c5c99b4e10031ee : SUCCESS in 3m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fc97cf1d5978442fa3534eba773aa624 : SUCCESS in 5m 54s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/47ac3d97f3f7448a97b3710765a1a1a9 : SUCCESS in 4m 32s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c516834087ab4b77b79e45b3187d4233 : SUCCESS in 4m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e6148ec13b7b4b9eb224ea694f8dc8ee : SUCCESS in 4m 52s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/af0e6269e3904c30ac5a734cfa24cafc : SUCCESS in 10m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7f17ad57166642328a2433936ff17278 : SUCCESS in 5m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ca00daba9bef4d94b1b9fe19bfac9b77 : SUCCESS in 4m 59s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/337a64b22acf4c1f870592eb8e51cb9c : SUCCESS in 4m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/077a46b85bbf419b8d945c72a7575778 : SUCCESS in 16m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a893a08f400a4514ad19b89a56c047bf : SUCCESS in 4m 29s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/20053e91e5334ab080b8e81c3b60035c : SUCCESS in 7m 57s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f22e9b6d11dc4135ac27f400da8b7a6b : SUCCESS in 7m 47s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fbfbdf4920c24af98893cc4d42ab9dd4 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5c3bce7abfa5474db7402be010aeb60c : SUCCESS in 8m 41s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a57d7fc5b78645d28290b2ef1d90aa77 : SUCCESS in 3m 46s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d7d4659f35664a919635648a60ea48a0 : SUCCESS in 4m 16s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2105b232fcc34da0ae874ce4b584792d : SUCCESS in 11m 19s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/78fc9592a0e14eb39451a081ad33c306 : SUCCESS in 18m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91f84865817f49cd88914efd46ab7022 : SUCCESS in 6m 04s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ddaaf39feaa94722ba7bee940b492b91 : SUCCESS in 12m 30s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/bf710e71c9514c9bae1eea92bbd227b6 : SUCCESS in 2m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7132d9e9e23040dc95e2978b9484f87e : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5da6dc11670745cf82be82fcea93a58b : SUCCESS in 22m 27s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6ce90187305f4acd86ebd56f70683801 : SUCCESS in 4m 45s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fd75d6a84bd14bd2836d93fe9d92cc70 : POST_FAILURE in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bf40b4982d14b0ca7e998e0203e0d09 : SUCCESS in 5m 03s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c1c572b5ce4b4810a337011c64a447de : SUCCESS in 12m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/131510efd9544c9285ca8defec86c0af : SUCCESS in 4m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56b54b323ff54e9182a333c4a058bb02 : SUCCESS in 36m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6a913e6b1aa4d37acbc948ac82f4c51 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e823c3f6cd15f2601701924ab024ae81d70e87eb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-15 04:37:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db7081bce1e7e1dd87f43ecf2fa99ac8eb8b7a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 06:19:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f01d2302224d41a9ee010ed8184b29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d3975ca2524189bc0465bd149efde0 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8954fbb6679a4cdbb534e8e9b7de3e4a : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/358ed8f0593c41a58569c68db5ba8f34 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d9a668602634418bdb45b23edf9c5a9 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d4e860efbe94732a4089a22114f8b3f : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5b2ffdf268a4f2f91ea11a8a2e0dee5 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24c370fb136c417f870e9bb0d5d29d9a : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/05cd230488004dd0a5f7391458baf283 : SUCCESS in 4m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5af3e2c51c8541b0a27a05f591fd8840 : SUCCESS in 3m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/365c9823459c4b349c181943a0b87409 : SUCCESS in 6m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f5d1cb8d5e1c4e038be76cb2de2242e7 : SUCCESS in 6m 03s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6f231c7cf6cc44bcb1d8516f0ce44237 : SUCCESS in 4m 02s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e79eab16c84d4b32b1e5f3a65d3ac24f : SUCCESS in 5m 08s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/d2bf5bd7b0954d74987ab18ede121cef : SUCCESS in 4m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/dc74f47b4c2b4723abcb945ef2b569d8 : SUCCESS in 7m 35s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d076b59403c1479f84a24fe9b746fa09 : SUCCESS in 5m 09s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ea7c8a41df064df895075d403803ff16 : SUCCESS in 3m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e7d016cf71124af7ad4627697c8ccf02 : SUCCESS in 6m 08s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d57dbf6da08a432bbb19c22c25bca260 : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93e0af1270324da49b7e6771834b2c9d : SUCCESS in 5m 51s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/05d2e26d2f984a2186b0bce9e7f12cc5 : SUCCESS in 5m 06s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/340141ae75c34195a188abb7af8e453d : SUCCESS in 10m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f32cc70a085a4e4fa699b13b8c023b93 : SUCCESS in 2m 12s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e2eec18b27da41ad9f2dbe7489b23dde : SUCCESS in 8m 57s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9db99177f75646cd91dbf2b37c306b31 : SUCCESS in 4m 20s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/1eb08869af2a4330a09e9bde71ac33fb : SUCCESS in 7m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d794ebfb7fbf4afabdd2092367a4fc92 : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7abb33272a014863a7a3f04cac0b97ff : SUCCESS in 15m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc926bb738c449848a82bfd076a651ef : SUCCESS in 6m 12s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f6fe9bfda93f4c349006ab9a4dc3be5d : SUCCESS in 14m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0db2d137fd014753bf956cd2e353d52e : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8cab59341e64897a291c272cd8a876e : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a65593f0e8b34354aa85d2e17753c525 : SUCCESS in 19m 43s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/56035ee6f6b74168a6ca2df3bec93c4f : SUCCESS in 7m 53s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/19cfdefdb0f0438293e692bca0345b89 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ceb2acd3a6f64a4290f44bd3ed8e83f8 : SUCCESS in 5m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/5b8b32a2107441539c10b75a1c6278c0 : SUCCESS in 11m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c8c0ba5f1d5842599d25b75b00c6e143 : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db2ea298e37f4824afd7800cf57141a7 : SUCCESS in 1h 02m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28858cd54e564fc3b3eca37029f5d319 : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ae4d07ba3b006c4d57ab5a1e0b7096ffb9ae0fce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-15 06:59:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8da3a75f96d77c43fd71ab299c656ccb5bbba839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 07:00:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f863b40c0e09489edbcc1d5c40bc4b859265d9f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 08:49:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c62b2dedbd54f97a551436eec8cdfde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb32d20539c4cf78e9e00be43caf010 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f616691896634c5fa367b7c55eab199c : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14281dd6dcc440af8ed650e4cedfb79b : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15e3faff88b4488ea7550d18418f274e : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb06962f88904972ba137934708a6964 : SUCCESS in 7m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b75ab393d9248198786a90c2b8f5392 : SUCCESS in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/caf99401c7ff4c25853636d1d1cce99b : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a9bb863a08e3461d854970b4e342832f : SUCCESS in 5m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/134e7cdad55d4ff28b02f84ee362750d : SUCCESS in 8m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/14fbf3cce7aa4256a501d3ac5062512e : SUCCESS in 3m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/43b3c97d9e46461cbc3ee938f00b3c7a : TIMED_OUT in 41m 12s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/50f127b215bd442abe60ec7b5636bcb0 : SUCCESS in 3m 48s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0808de52f5da4254aa51d45a57312387 : SUCCESS in 2m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/cac31c006fc046aabeade096b2646100 : SUCCESS in 6m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3dc04c7925864380b43505e60304fc87 : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0de9829b5126453ba764650da9267dd3 : SUCCESS in 17m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/125c1fbd389041be919acb446c7b38a6 : SUCCESS in 5m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d1c553945c6f429b99b45a098cd249c3 : SUCCESS in 8m 16s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a4ade81464864525a4dcf560c51ebbbc : SUCCESS in 1m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0a4443d68a58457c8a37389905320a0a : SUCCESS in 7m 28s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/bd7080dfe0cf42bd97531292833041a0 : SUCCESS in 25m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9e248c00476a42a8b10d432c2c63e02a : SUCCESS in 6m 40s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/872850832adf416f998f611a72763867 : SUCCESS in 12m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/70fcb3f63e7f4ceda3b911f8bd54652e : SUCCESS in 2m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d28f8f7c69c49539302fb88d9f81de1 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1413ebd9af4749d8b8fcc870f63c8ab1 : SUCCESS in 25m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/702abfd41e8a4f6aa8e2a09fbe843d4e : SUCCESS in 4m 18s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/99e2cb5186964212bf5f7ed5b68080ec : SUCCESS in 3m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a265e4d7a5134a71b5350f297e0be6d8 : SUCCESS in 5m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0adf7b404a1045199185e2e58928caa6 : SUCCESS in 14m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/66d598eb1b53461e839e2015005ab128 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc94c6398feb47e9819df1fee7b210eb : SUCCESS in 36m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb75fbc4bcfa40009ee436068674502e : SUCCESS in 1h 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"464ed2205d10079d6d60f16bee9333fbcd295262","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-15 12:01:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"111227dcf42888c06c0aa33255cd238b756a9260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 13:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa0ad524b924fc1af0ad6f82aaaa065\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c1a35bf2b74b60996c6f3e979d0434 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd07c7b30894d32abf7515526b9118a : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a4f6221938848fcb1cac45ec60c37ce : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee1e8cded30b4ac19aa15c54554915fa : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b6f2e299fc4402c873e4524a75a0ab8 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5c5c6c1b7b745eaabbe21bc4c938906 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8b14b83589e442db5c3ba77f635dbc8 : SUCCESS in 3m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e4822bbb0f8c4c179a0a88b68ed37482 : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4001d9b99b3a42f38f071ca7040a347f : SUCCESS in 5m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ff72505dcaea457eb5fe7de91d3703f5 : SUCCESS in 6m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b41e0cf773c54bda8e7e9892795ebbb5 : SUCCESS in 5m 35s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5b4cd06e07dc41d3b3a13952452b7013 : SUCCESS in 7m 34s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/7e7b794ace864d78af587120db853a60 : SUCCESS in 5m 02s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ec9464699e0043ec8e7185a568358f58 : SUCCESS in 4m 35s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8108f2f726fd4ac582705a0a265f8126 : SUCCESS in 9m 54s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/93ea8d3b1869403cae7948add201e7c7 : SUCCESS in 5m 11s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a9e5d64497ef4801aa38e9b4d2f4eae8 : SUCCESS in 5m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9bd35b340f324558995c4b25751da81c : SUCCESS in 7m 01s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/318e4aee1b174ebd815bbf20f9436193 : SUCCESS in 14m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66e904e809ab4f7cb4820723c7b15c77 : SUCCESS in 6m 09s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5cb7715e2ca245d2bb42e12ad3df31f2 : SUCCESS in 9m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1a342472dace49acb47ee6359de23cac : SUCCESS in 11m 03s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/cc4ea478b9a84abf9e70ee0e4f240d82 : SUCCESS in 2m 50s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7b4c0c7cea7a45a8a49bb710704aab10 : SUCCESS in 7m 34s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/057d316ddb41450c93eaf8dacd6058c4 : SUCCESS in 3m 02s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/87f10e7ad4b54c8b9ba36f880a38a748 : SUCCESS in 6m 52s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/3ae6bde88b8746b3bf9584ec0adbb0c9 : SUCCESS in 9m 39s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4f3f0e21985a45a99f3337a80330a390 : SUCCESS in 27m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/58f74b0211364a74b0bbc1042301b64a : SUCCESS in 7m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8d2b098bd69d40bab10fb22ab20117c7 : SUCCESS in 13m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/908e08ed3dfd464aa64b606bd5fe98f9 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29143292932c4e8c89a32bf80bc6a98e : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff5b2085b06241638809ef763a4d268b : SUCCESS in 25m 48s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2a0dee9bfa4f4279a341d5b5d41e97fa : SUCCESS in 5m 48s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b77cf7b1e20d46b4a338846645e58648 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/363264dbfb874102adcdeb76dfcb906f : SUCCESS in 6m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/367be0c5db894b99b13662e04446bb4f : SUCCESS in 18m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79f6dc7dab1249f8a2f9ed729ffa5fef : SUCCESS in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde6a93ae11c403da47b76ba2bc6653b : SUCCESS in 1h 04m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de0b2a79e1d4484cb7750db3de43d596 : SUCCESS in 1h 49m 05s","accounts_in_message":[],"_revision_number":1},{"id":"164cba30b393fb62b78aaa67dfd5c8e572646812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 13:59:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"389d9131eb5052c8073cf1f20537c9a306999fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:44:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c192685d21b849528d23ecb194b3c136\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13576f387c154a22907af715d42ba54e : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e030b3afa1fc4b4194d1613f09d55cc1 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a7373c2ace240eeafed4ace6d242702 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f7bff5cd0aa4ab083e58470dbdb19d4 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a3799499d054d5d8f6566f48482c461 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7044b926e3745e98789cac56054ae94 : SUCCESS in 3m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/01ef316c0cf6422899b10c82e1033416 : SUCCESS in 13m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/660cab6847ca47bfac69c03ebbbe386a : SUCCESS in 3m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3bac19a21bcf4d3f9c810bf9f779fd44 : SUCCESS in 3m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7857e87bc55d47d0ac1461bc095d986e : SUCCESS in 3m 18s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/142ce07a1a8f46ba85405e3cb9544aa4 : SUCCESS in 11m 40s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/760a6aa64e5f44118bdb7dac0e44f050 : SUCCESS in 4m 04s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/19a2fc935b774e19b02e80e2ee5af131 : SUCCESS in 5m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/73c32f592bae459f8f68db0c01421b5c : SUCCESS in 6m 35s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/bed6a5b4faa0442c98d2070ce85a8b68 : SUCCESS in 16m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae50ed7ede4b42618f31dcb6b5d6ee40 : SUCCESS in 8m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/642fa32c9f4a42fca94b89ce19e99e2b : SUCCESS in 8m 02s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0c0d18a5201544a597c80ece15196999 : SUCCESS in 11m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5b76bfdec4814180a12d4187f500af12 : SUCCESS in 4m 05s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b4033e0995f24daf8c9984ef1057a7ee : SUCCESS in 7m 29s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c395011e63994389baa0f47a18b8a327 : SUCCESS in 21m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/191fe7ae35264c979f4af5d8601cff61 : SUCCESS in 4m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/56b3f2ff3b944acf9f8bf73154b533bf : SUCCESS in 14m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b6c719c3b2b444a4954e4e324d60b5a7 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e94059c3fb6346fa9e8c601c8e100bf9 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7af3c19b246344f7b42315ca4152ab5c : SUCCESS in 27m 28s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e56e361545fd4bae9ac4c2f6a1062237 : SUCCESS in 6m 56s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4bf91206112e43fda357412b08a653b6 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5911eae1f6a8444eab4a5f2f919e2631 : SUCCESS in 7m 46s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/89f5be3c94584f4390a87af77af0c05b : SUCCESS in 13m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ba81961018bf41b499ffafcdab6dfbbf : SUCCESS in 5m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df7518b994fb425eb810d5740703dbe8 : SUCCESS in 55m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/236aa84bf4004ad6a897aa8ccfc42ce4 : SUCCESS in 1h 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0d9961d1fbe37eee0dc594b91c88c56773abf56c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:44:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e37628c7080b47d3af100089bcb8ea7bb2d73c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:44:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0b0e83e85b4581845ede038e46fb0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8d7e0d72594ed5bbf1367fcb3bbba3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4bb697e1d700309d88eebd8433e00da6f0eba7c","revisions":{"f4bb697e1d700309d88eebd8433e00da6f0eba7c":{"kind":"REWORK","_number":1,"created":"2026-04-14 12:20:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/984552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/984552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/984552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/984552/1"}}},"commit":{"parents":[{"commit":"84e467a02fc484ca063ae383ab95d04f530090fb","subject":"update constraint for openstacksdk to new release 4.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84e467a02fc484ca063ae383ab95d04f530090fb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-14 12:10:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-14 12:10:29.000000000","tz":0},"subject":"update constraint for osc-lib to new release 4.5.0","message":"update constraint for osc-lib to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ib82aeb4129030ca715a067fc54027002d073d57a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie4428c5ee444a51eac22dcd935172b2160b5b56c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4bb697e1d700309d88eebd8433e00da6f0eba7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4bb697e1d700309d88eebd8433e00da6f0eba7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
