)]}'
{"id":"openstack%2Frequirements~985353","triplet_id":"openstack%2Frequirements~master~Ib8bd4b27a4a1570c0998c7a2a3002b75bbaabb3f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib8bd4b27a4a1570c0998c7a2a3002b75bbaabb3f","subject":"update constraint for oslo.versionedobjects to new release 3.10.0","status":"ABANDONED","created":"2026-04-20 10:11:35.000000000","updated":"2026-04-22 08:18:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"37c3bf8c56d347832166f510da5abfc135756fd7","_number":985353,"virtual_id_number":985353,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-20 12:10:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2026-04-21 13:40:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 12:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-20 13:32:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-04-21 13:40:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"36165d24aa5e26c444fe1db295b10f877026900b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 10:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1599a119fe254185b105a458995e51f1b9787d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:10:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/598bc0c42b2c4e3ea0f6748cb3876ed4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a8101c314349f9abaa91d42743f31c : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4f692d39eb47959c3bf885783fb32e : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e55f126110b439bbb3f4a19f37ab13c : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a15615469054d8c95076d7d252e33e1 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/270355cc8ec14f3f93d08fd8335c1efb : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c66c748c1254519a02ac99421164cbe : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e75a51131884474986016f38f31c544 : SUCCESS in 3m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3bbd52b858e94bbfb52b36aa6323a785 : SUCCESS in 5m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d0c74aaacdbe4c85937551489afeb134 : SUCCESS in 5m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e7a71c1958c5463791dfd1fffa6fd169 : SUCCESS in 6m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad6926cbf4b6404dbc447efe53f4bf61 : SUCCESS in 5m 53s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a6be145e91d14603843b4675a94f8e30 : SUCCESS in 5m 24s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6a60f8a263eb417983d07bef04be43bf : SUCCESS in 5m 16s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/11a2eec168e14c1f855a3e5edddcb2e6 : SUCCESS in 5m 22s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/768775356d454c2eac500643613db8aa : SUCCESS in 11m 49s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4a07af520a344de6a38170d56e47e15f : SUCCESS in 3m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/19d4c2017feb4d859d3d44fb6cf841c0 : SUCCESS in 5m 20s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9222331c56d748aa8c4c0315ffde199c : SUCCESS in 6m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/408b84c2173e4d338d3cfd420c88cd02 : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e2f0282e6364db2b92c6d67f2acdeae : SUCCESS in 7m 30s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/22def0e799f84594a611a0d80be9fac6 : FAILURE in 8m 08s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/10b02c8b0d4e42c8b1d3877553c19bc6 : SUCCESS in 9m 53s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/01f494d8c9ac491cbdc45f9f7920eba4 : SUCCESS in 2m 29s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b7899b24e2a747a4a7cb6a558d397420 : SUCCESS in 8m 30s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d1fe02353a8542be843fd8bb91189ec1 : FAILURE in 5m 23s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/fc1adcb98f114c8fbeb3af6e486a1fec : SUCCESS in 5m 23s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bd1b09b2af364f8381479675c5d77b67 : SUCCESS in 11m 19s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2393c0a50f674623aeed528013ec6cb0 : SUCCESS in 28m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/effcb65f135e45679c440f5876ebe316 : SUCCESS in 7m 48s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/619ae7117618437984a240778d35cb10 : FAILURE in 14m 30s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/afa6615353a94cd99a554f2ca6c3c3ab : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd5b966964b04a1798120d52a2e3297f : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03fa9365adf4498ea98f4d3efc3999fc : FAILURE in 15m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/9f1ffdaaa7da43ff829cc44f6a30c7ad : SUCCESS in 6m 14s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8b9335098f79464994b29a436285fad6 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08a1464d52ce435d94b17092c0d67aaa : SUCCESS in 7m 41s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/90a5ae8a801c4d30acca6a9e5451e3f3 : SUCCESS in 17m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/06cb968acca24c25b7d9240a5a9871ed : FAILURE in 8m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7741a5405e594cf383ad613992a0ad22 : SUCCESS in 1h 14m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92c0b6a5a0614ac28ee064f592a49b86 : SUCCESS in 1h 41m 51s","accounts_in_message":[],"_revision_number":1},{"id":"75fc99ae8ee0af6c4382f32f50fb53006beabeb9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-20 13:32:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0990930e9d672f88fe95685bc02e106ec455b3d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-21 13:40:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a23cc2ed272751f23a698983efa703fad4a73d7a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-21 13:40:42.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"37c3bf8c56d347832166f510da5abfc135756fd7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 08:18:14.000000000","message":"Abandoned\n\nsuperceded by 3.10.1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e43bd4c55d270836320c13b427f7a62c11edccda","revisions":{"e43bd4c55d270836320c13b427f7a62c11edccda":{"kind":"REWORK","_number":1,"created":"2026-04-20 10:11:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/985353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/985353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/985353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/985353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/985353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/985353/1"}}},"commit":{"parents":[{"commit":"954a2cc76030c7154969416f0cdfe1568f986c37","subject":"Merge \"Add the Sphinx copy-button\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/954a2cc76030c7154969416f0cdfe1568f986c37"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 10:01:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 10:01:34.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.10.0","message":"update constraint for oslo.versionedobjects to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I3f86f1d1a2df94eba08d71852997f10b98c0929b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib8bd4b27a4a1570c0998c7a2a3002b75bbaabb3f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e43bd4c55d270836320c13b427f7a62c11edccda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e43bd4c55d270836320c13b427f7a62c11edccda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
