)]}'
{"id":"openstack%2Frequirements~985707","triplet_id":"openstack%2Frequirements~master~I4a33f4cd56a62cfe75abe8cd62f0686a05a1c715","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a33f4cd56a62cfe75abe8cd62f0686a05a1c715","subject":"update constraint for oslo.versionedobjects to new release 3.10.1","status":"ABANDONED","created":"2026-04-21 15:52:28.000000000","updated":"2026-04-22 12:34:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b93cfad9134bc0f76d0bb80859a0aae8c63fa49","_number":985707,"virtual_id_number":985707,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-21 17:46:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 17:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efcee32a681abe2177355fd842940edf24555866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-21 15:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"573f93dff16f81ac7a4a007eb4472dcac3ca69a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 17:46:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/219676e2fb3f4aca9e349059e82917cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab56727df7a493983600e471a454da0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085350c3e32a4fee8ac9a449c1309122 : SUCCESS in 2m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77d100cef9ae49d48b0910462747c08c : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/393b5c3e82964d3a9ec1f23d9f2c0c47 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11fbccf7f06d40898d1e5e1248ae763c : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49a431ad55764c7e9c8d566f8c692524 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8620e317e0664cff9b2d209c0c180488 : SUCCESS in 2m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cb9e2c1ddda24c13879c4cec56fd9c1e : SUCCESS in 4m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1baf7b54e87a4be093d849c997e7a13f : SUCCESS in 5m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b8094ac92b374ab0b37ca153bbc83831 : SUCCESS in 8m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/10f5762eb32c405b97c938c8b4e327c8 : SUCCESS in 5m 36s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/816b9365b2f54a3d86c7c9c4b4aa6cde : SUCCESS in 4m 20s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/692c9d1438f54887b69a95142b0765b5 : SUCCESS in 3m 00s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/cb1038d0542e45b0bd7c7934179bc908 : SUCCESS in 5m 22s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5ff8f01099c74e5fab68452263abb4c5 : TIMED_OUT in 40m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9044d877a14848f9ac4d60de8f4e2895 : SUCCESS in 5m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4e9b46614ea14a0ea1643408a12ce9c7 : SUCCESS in 5m 12s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2e967edc9d1345aba6e9903a6a178274 : SUCCESS in 6m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0317a5833a864783910da7378ef05a3b : SUCCESS in 15m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09fad9307c5b4f0085b1fe3de9dc7823 : SUCCESS in 7m 50s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/25c5235def9a47abbefec6af2a1fb269 : SUCCESS in 8m 33s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/640d3d6613b94fa2b0bab7a3bdceef8f : SUCCESS in 12m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5599cf5cfe80437da81a6f3c4fcc6cdc : SUCCESS in 2m 22s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/374b414f16ce4e059368b7c9ba79430e : SUCCESS in 9m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/95c10e7cecc64a459324bd8f9ae026a9 : FAILURE in 4m 18s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f2ec949069bc470fbd3d0a708570234e : SUCCESS in 5m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/da122d8d86994a6589d00d0abffe4c1c : SUCCESS in 10m 01s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/535aab202c5b41048767661d2e298567 : SUCCESS in 24m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17287580edc54151918e7f41474d4251 : SUCCESS in 6m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/01dd91781a024bc8b822bea1dbc5fed1 : SUCCESS in 13m 04s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e58d6eeabf6847fe8752f872b2873e7d : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98fcc148f23d411a8dcaf6dc5d3421d3 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a8e220ef0224314a8de54552c7783d2 : SUCCESS in 25m 28s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/8ad90156726d4837a905d7aff16d7ab2 : SUCCESS in 5m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e313a4f06c494842ab89f6c2ea8aef97 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31ef78d606c54d88bdd0b96b3451a365 : SUCCESS in 6m 45s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1c4dcf2b9a3a4803adf0095a3383df3a : SUCCESS in 19m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5a124b827b224a9fabdd572874143d7c : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/901a356d0c674f6781ecee039a7a26dc : SUCCESS in 1h 05m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6aea62deb3934104812b12f00aaf2027 : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1b93cfad9134bc0f76d0bb80859a0aae8c63fa49","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 12:34:39.000000000","message":"Abandoned\n\nthis is still missing a fix for magnum, will be superceded by 3.10.2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c0d8a1e044cf8b410369b1bca369d96911901f9","revisions":{"0c0d8a1e044cf8b410369b1bca369d96911901f9":{"kind":"REWORK","_number":1,"created":"2026-04-21 15:52:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/985707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/985707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/985707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/985707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/985707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/985707/1"}}},"commit":{"parents":[{"commit":"954a2cc76030c7154969416f0cdfe1568f986c37","subject":"Merge \"Add the Sphinx copy-button\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/954a2cc76030c7154969416f0cdfe1568f986c37"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-21 15:42:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-21 15:42:27.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.10.1","message":"update constraint for oslo.versionedobjects to new release 3.10.1\n\nmeta: version: 3.10.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I3115bfb3ca2ac16bf012f9026ba8e7028d017ab1\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I4a33f4cd56a62cfe75abe8cd62f0686a05a1c715\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0d8a1e044cf8b410369b1bca369d96911901f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0d8a1e044cf8b410369b1bca369d96911901f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
