)]}'
{"id":"openstack%2Frequirements~985811","triplet_id":"openstack%2Frequirements~master~I8e091e70829078b551789567cfb5f5570dc589e4","project":"openstack/requirements","branch":"master","topic":"new-release-master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-22 15:20:45.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I8e091e70829078b551789567cfb5f5570dc589e4","subject":"update constraint for oslo.concurrency to new release 7.4.1","status":"MERGED","created":"2026-04-22 12:41:09.000000000","updated":"2026-04-22 16:41:20.000000000","submitted":"2026-04-22 16:40:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985811-new-release-master","cherry_pick_of_change":984381,"cherry_pick_of_patch_set":1,"meta_rev_id":"398927204d25e925027af92f15e9ed37c00f4f4f","_number":985811,"virtual_id_number":985811,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 16:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-22 15:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-22 14:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-22 15:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 12:41:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-22 14:30:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-04-22 14:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"272fe74acde8881634c2544c6dca973c0a70170c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 12:41:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"09a6253f69604577fd0356fd97dbaf98b95188ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-22 14:30:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2a75e6e4039c3f59661a66c8323f59571e8a801a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-22 14:30:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e86f777e461e246f7de9defcdb8dd8c24162a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 14:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d0beaf4f1047d2b250692b0e61bde9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5276a5c9e54774a63054916e3238d9 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9366a0dae92845d2a8e3326d080bcc15 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24e05531dcb54b31a8801a1cc4554619 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8d87b0197484c2e914046c722ca0254 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/882c16bfdb9a4b488e3bfc55b76305b7 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ea5ee8c06234024b9eba2104b7ccd74 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/955ba36a96f54fab8f55c12ccdd1505b : SUCCESS in 2m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/194118e63eb345d4b5630707fc8235a4 : SUCCESS in 4m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4c7f442232714cdd94669e3338c55e4e : SUCCESS in 6m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b23c8551be754709b0e86e91ca7ed098 : SUCCESS in 4m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f19b5eb939014ec98c45ba393fea0d83 : SUCCESS in 5m 29s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/865754d1d4724feaa5639bcea6e86d9e : SUCCESS in 4m 09s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/42a6bd1513b4438b9ff7ca863021ea8b : SUCCESS in 5m 43s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ecc1fdd2df75468b9f441b269a9f79ad : SUCCESS in 4m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/38e337f2956646509ad49a230a60eeb3 : SUCCESS in 9m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ab7c63469c1a4c60919948eac05d8e1d : SUCCESS in 5m 28s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/01b7ea4e88b24491a65a198155879944 : SUCCESS in 5m 10s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/79b328b2db40498ba3b4fcd3e03eddae : SUCCESS in 4m 16s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/224a208e97fa4d73a51d2d1e59ce97b2 : SUCCESS in 15m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8db5ccc68c20420db914373d91949c5d : SUCCESS in 8m 05s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b78e764cd2d14cccbccf5fff3508f3aa : SUCCESS in 7m 01s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4c46b53db8f949cb83eb174fe8bd484f : SUCCESS in 12m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a02eb7fd2c66435fa8467dc1145160d9 : SUCCESS in 1m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/42d37bbf1a1c408daa8bfe1ee61843b9 : SUCCESS in 8m 36s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/72cdf82d5f5942f3baaf3a0e3005c781 : SUCCESS in 6m 23s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/6c779191dfc94894b2df1faaf51de0aa : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ba09c491fa3f4375b8dd0ade9ea3c8c1 : SUCCESS in 9m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/86e73487d4394e9790297d969d9645ed : SUCCESS in 19m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5768d25a801a41a2bc6187550b2131bc : SUCCESS in 7m 27s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2a4daa2bf38f4aafad576c5782b0580d : SUCCESS in 14m 02s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ab5a481fa1ec4ab7a5b49a13402ea938 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0acd00afe1e84335892cb80be0944147 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7d899d98ee247c1a5a53a41c0c9803e : SUCCESS in 25m 01s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0f4bb1b26c6740e6b1b91b1629eab87b : SUCCESS in 5m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fa76011de1be4c93bc79f2607367cc82 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fea732e5c014696963ba5b64cb2454b : SUCCESS in 7m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/49060b56932e47de92c2ab3ddc989c14 : SUCCESS in 12m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d5b1642f601a417a885083a563f0bc78 : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8de9f601f74d479e87171f1d962be6dd : SUCCESS in 1h 04m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d7973f35d7c40fba5bf191b58106094 : SUCCESS in 1h 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ef71ccf38faa9056ffb3503a5c6cc1cf995e0873","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-22 15:20:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"76e86ce7cead547a684017cf4a3db17617ea6b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 15:21:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"601f5c655d07361b0a6db02c25d785b4987524f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:40:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490e3616ad494dc48c10ae23f3415e7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afbb81bbff54bc7a3ac78e0bc6aef8e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c45b265872465295a946fd7520b93d : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8703b484a4d4b4eae4b22aacc67e2e8 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56dcbc45a35a48fd8b4ad48df178fa83 : SUCCESS in 2m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70d7baa5d5494a20bd87ecd1f8073eed : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0ae6f6393a9403790aa68d9462f4489 : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/490ec97bf154409f8d4305e9eb85c6fb : SUCCESS in 4m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84f36fbe5dbb40d48e9d0c7233adcf9c : SUCCESS in 3m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/59c4a10a95d44bdb984908ca92f54aad : SUCCESS in 6m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/488090d5122347c78fa2e8afded6db48 : SUCCESS in 5m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/7af3f44d9d8c4bddae81725bcbd5d5af : SUCCESS in 12m 08s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/330ac969fa22492aa39a2e7688f451ad : SUCCESS in 5m 03s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/bf8c1e0732d04523b81813bc4ab558f2 : SUCCESS in 5m 07s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/4790bf2c1f894bb985298d16499f23e7 : SUCCESS in 7m 43s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/188dd5c9cd81440e877ac2f8a0887da4 : SUCCESS in 16m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f0be2cd9f9b46e5820222210e9a5887 : SUCCESS in 6m 39s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/45a36001eb0a47b2a9ee0b95a34dd70b : SUCCESS in 7m 47s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a0871febb7514f7a9b1e16935f851e52 : SUCCESS in 11m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ad2b7e1dccc04cd8b2dcefc22af73e4c : SUCCESS in 2m 50s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/837a77ee77f347f0b835d4c1f871c92d : SUCCESS in 9m 52s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/052b9ed3887b48a39d320bdf656655d2 : SUCCESS in 23m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42281b3dd6524c6ea34d4c1381628244 : SUCCESS in 4m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6a8171c068d74ca5a56403d8f78c04f7 : SUCCESS in 14m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a4bd715f70874dacb0ff4b1a94f2b324 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa95a607692e4fb39e51c384cd0bff21 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d10a6e43d8a46c3819596a370f21fe0 : SUCCESS in 24m 07s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/dd71f01ecbb44cc6ab26b424ce21b241 : SUCCESS in 4m 07s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/1ebb4404eea64a95af59ee966840388b : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6b9b183f944418fa4f067bac1efe2fd : SUCCESS in 7m 27s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/648929de78b24b28b521db18421192fa : SUCCESS in 20m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/071f7f1922034758af12ac9f88605551 : SUCCESS in 6m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ed753deaea841139a96640963f86ea6 : SUCCESS in 1h 13m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e64784d9598467c8f1b902f39dceb89 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0c2145084eb2dc32a4e8c45102405537cbca06e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:40:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"398927204d25e925027af92f15e9ed37c00f4f4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 16:41:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6e0e6f83c54a7a851e35028ee06283\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a753fb8427ca4037b35895eb072b8090 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da1afedda8b79d2e3e9b8869c615b8cb1cb14fb9","revisions":{"da1afedda8b79d2e3e9b8869c615b8cb1cb14fb9":{"kind":"REWORK","_number":1,"created":"2026-04-22 12:41:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/11/985811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/985811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/985811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/985811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/985811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/985811/1"}}},"commit":{"parents":[{"commit":"e30ba7d5ef830ac9b96382e5ee238981604b18c6","subject":"Merge \"update constraint for neutron-lib to new release 3.25.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e30ba7d5ef830ac9b96382e5ee238981604b18c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-13 12:41:57.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2026-04-22 12:41:09.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.4.1","message":"update constraint for oslo.concurrency to new release 7.4.1\n\nrelease was made on 2026.1, but since it is the newest version overall, it needs to be bumped on the master branch, too\n\nmeta: version: 7.4.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ib4f005504e94e9b5ef220b3d1226d630ec919865\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8e091e70829078b551789567cfb5f5570dc589e4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n(cherry picked from commit 7e07cb254ee7786723be7030ff3fd438e2960e2b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da1afedda8b79d2e3e9b8869c615b8cb1cb14fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da1afedda8b79d2e3e9b8869c615b8cb1cb14fb9"}]},"branch":"refs/heads/master","conflicts":{"ours":"e30ba7d5ef830ac9b96382e5ee238981604b18c6","theirs":"7e07cb254ee7786723be7030ff3fd438e2960e2b","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
