)]}'
{"id":"openstack%2Frequirements~986068","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Icfe0dcdf439e5d69c5ee3ea75171b506b6fdd725","project":"openstack/requirements","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-24 19:43:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfe0dcdf439e5d69c5ee3ea75171b506b6fdd725","subject":"[stable-only] Fix py38- and py39-check-uc job","status":"MERGED","created":"2026-04-24 10:19:04.000000000","updated":"2026-04-24 19:43:42.000000000","submitted":"2026-04-24 19:43:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986068","meta_rev_id":"395a846447497a7707df42e54fba9cecb3ea9afe","_number":986068,"virtual_id_number":986068,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-24 19:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-24 15:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-24 15:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 12:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 15:55:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"485cd1e468a92237bcca83adbcee0f4f0977228c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-24 10:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fe0f673cdaa4b7c70367414413cfada1787a9b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 12:14:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f6439f28363440493ef98bdcb917cb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5685a47fcdd3430cb5ad3a90e90e84b3 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae4967fe9c2b47b69cdae5ad1d375c39 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1226e2d4854e4c13ada5b3f51dd6499a : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0653fb5e57494413b5d755207497d6a1 : FAILURE in 4m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f29e12bf4854a4cbeb30bd7e9d9d5b5 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29a2828209514d65af240a6e88cafc14 : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20dda90c6c85457db12e99262e8c23b2 : FAILURE in 47m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37d56b5dc77c43d98937b70de238abed : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"af5fba66c1a3e4b57d7b8ea4d3abef8d8c65a1d7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-24 12:17:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1163b715bb3c35668ca076ef63cee2fcd9031fb6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-24 12:32:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b7d26a564182d0d844177998cd6a1b4a4daf376d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-24 13:35:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"2df41a37915fe57edb290eb7dac7b198bc700906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 14:08:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b0b7058f39469da8114e9b0972b3f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc0accaa5fb4cf9aa81d5bf503fa41c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b089f78fe64939b68ec00b6b98be93 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fe96ab6136847bda8858556ecfb0a48 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5973081f3bf44172b7422dabfae7abf5 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dbf4323c7fe4a7cbb408a2cfeca14b4 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c6c901d802c442496868f585a852243 : SUCCESS in 4m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/785e526ae4b849f0893cf159bee77f51 : SUCCESS in 58m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32bfe1b1e6af4e1b95da7e6e428a2653 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"393038becc6c295cda67471b02a4cabf9170f5ec","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-24 15:55:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03d390ec0fd8571241e0b9716793f8dbaac97d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 15:56:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6309f6814d47b489c0ac34a40621c7dd58d1125e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 19:43:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311a23607d03476bacb381e262b328de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e23d9366b9d48bbaa0854126019606d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/744a31b9e94a4776b5e405d1cd8463ea : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a55558bf3f048a39ccfe885e8ee16e0 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbc99d3bb9484040b06c78889e7d0549 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9f194db620549a3856275cbb678088e : SUCCESS in 4m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7847c1c3aec0423ba7d548b72689fcd3 : SUCCESS in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8a824b7359a4a11b23851b11b3b7c16 : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":2},{"id":"395a846447497a7707df42e54fba9cecb3ea9afe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 19:43:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f19c048af89341defa6b891e0709f04aac358de","revisions":{"4f6c79f8d424b0887ec43800c0eeb150a5fb0d53":{"kind":"REWORK","_number":1,"created":"2026-04-24 10:19:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/986068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/986068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/986068/1"}}},"commit":{"parents":[{"commit":"930177f70bf0ae58050ad82a558c464e7438e898","subject":"Merge \"stable-only: drop bionic wheel cache job\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/930177f70bf0ae58050ad82a558c464e7438e898"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-24 10:15:31.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-24 10:18:48.000000000","tz":120},"subject":"[stable-only] Fix py38- and py39-check-uc jobs","message":"[stable-only] Fix py38- and py39-check-uc jobs\n\nNow that the tools/pip.sh helper was added as a workaround to fix build\ntime errors, the mentioned two jobs fail with \u0027tools/pip.sh is not\nallowed, use allowlist_externals to allow it\u0027 errors. This patch\nextends the allowlist_externals of these two jobs.\n\nChange-Id: Icfe0dcdf439e5d69c5ee3ea75171b506b6fdd725\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f6c79f8d424b0887ec43800c0eeb150a5fb0d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f6c79f8d424b0887ec43800c0eeb150a5fb0d53"}]},"branch":"refs/heads/stable/2024.2"},"9f19c048af89341defa6b891e0709f04aac358de":{"kind":"REWORK","_number":2,"created":"2026-04-24 12:32:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/986068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/986068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/986068/2"}}},"commit":{"parents":[{"commit":"930177f70bf0ae58050ad82a558c464e7438e898","subject":"Merge \"stable-only: drop bionic wheel cache job\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/930177f70bf0ae58050ad82a558c464e7438e898"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-24 10:15:31.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-24 12:31:26.000000000","tz":120},"subject":"[stable-only] Fix py38- and py39-check-uc job","message":"[stable-only] Fix py38- and py39-check-uc job\n\nNow that the tools/pip.sh helper was added as a workaround to fix build\ntime errors, the mentioned job fails with \u0027tools/pip.sh is not allowed,\nuse allowlist_externals to allow it\u0027 errors. This patch extends the\nallowlist_externals of py39-check-uc job and removes the tools/pip.sh\ncall from py38-check-uc job, as that is not needed there.\n\nChange-Id: Icfe0dcdf439e5d69c5ee3ea75171b506b6fdd725\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f19c048af89341defa6b891e0709f04aac358de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f19c048af89341defa6b891e0709f04aac358de"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
