)]}'
{"id":"openstack%2Frequirements~986249","triplet_id":"openstack%2Frequirements~master~Ib909ecfebb541e306444f4ed6483f5d48dbe807b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-27 08:14:00.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ib909ecfebb541e306444f4ed6483f5d48dbe807b","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-04-25 08:28:23.000000000","updated":"2026-04-27 17:23:56.000000000","submitted":"2026-04-27 17:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986249-openstack/requirements/constraints","meta_rev_id":"f0c1b5e77f9aada1ce55ea5c60f6d265b1d6c199","_number":986249,"virtual_id_number":986249,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 17:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-27 16:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-27 16:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-25 10:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 08:14:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-27 16:01:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"13769e6ac5f27205fe99232380fc780f79bac1a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-25 08:28:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"631a1da9f8a70081507bbe88469cffc84984a13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 10:12:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d19098f4f30f4be1a84d5c56d100b1dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf563d79dd143768f659cacc7fa87ed : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f458905c1a748909cb94efb667af14e : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b149ea9957db40bd8a98365736600ca4 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8795cc36c1244b89f75d52a727fcace : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c8b9a11771649779f0cbb9a7c8ea86d : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbca63af8d814fd8998b521abd7f5f9c : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d834aedda76140188f27451008658811 : SUCCESS in 3m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7f63a2b1f5444eaf86dfc32c4970d125 : SUCCESS in 5m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f844e353c1604f609c9dcb56cee10015 : FAILURE in 2m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f30cec0cb5824c1a850accb857432cae : FAILURE in 2m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f8970de737914c528b271b1d0c57ccbd : FAILURE in 2m 51s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/eba0ecfe9fb04fda89e320d974c66e82 : SUCCESS in 4m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/1aadf9d8061e4d9c866dd7736f0fc5df : SUCCESS in 4m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b7d81d0637e447f8bc992512c2c89fc8 : FAILURE in 4m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/eed16b1360a04c299d1f3658c6934d17 : SUCCESS in 11m 11s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/394034d856664ba1aca256353df83fb9 : SUCCESS in 4m 53s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/52f374a8150f4e85be00522db4247245 : SUCCESS in 3m 22s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/35ad3b87a1f54779888d95b4044a6575 : SUCCESS in 7m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0db027916741431193c1aca9d47a0ebe : SUCCESS in 14m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0052a17392f4030a0f01fb5504836d4 : SUCCESS in 7m 12s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e39f02371c714919a12e7f381fead0d2 : SUCCESS in 4m 01s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/68557bc599994eceb2a0b73fac4d41fd : SUCCESS in 10m 28s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5f91dc48b1634bcda16c8f1474397a6d : SUCCESS in 3m 40s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/dfccc365c2034135a9a168e4848d921b : SUCCESS in 8m 06s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9f85a17fad054ecca0e8bcf31c221db7 : SUCCESS in 2m 35s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7dbafea895464b689499b41ec8b241f2 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/34b715f7ea7346a7a1717450c529d89d : SUCCESS in 9m 39s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a3a4c9a9a11f419685d165450313cfc6 : SUCCESS in 18m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bf645192a2e64af18bc33c3fdd28d057 : SUCCESS in 3m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0425699a48cc4177930c8522df2fef36 : SUCCESS in 9m 23s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/180de6efc2914f74b3da68d9e919484e : SUCCESS in 2m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/519b3305a5eb4e05a0abbe6273fb7547 : SUCCESS in 2m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac9704dced824a4794a720a1a751a44f : SUCCESS in 23m 26s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/97d5036ee5e24b8d8b98879fdf41c474 : SUCCESS in 2m 59s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/1bedbbe03e204737814e59250549bcaf : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/930ef8f4f5384a05a1b674174e62d332 : SUCCESS in 3m 47s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/05d5da3b4adc42618cc2ba006803f342 : SUCCESS in 7m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2f560f0dd0994f239a057bc24406d81b : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c0a37845f642fe87c1a86021b79dc2 : SUCCESS in 59m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8a79e987e064be7996c54e7941781eb : SUCCESS in 1h 43m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2c67ecb9b4f59c7b32d2162d0721f7824e28a5c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-26 09:44:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"560e5f4b2d1b0ce372b2e46c400d9ae60702aae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 11:31:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/679caad4ae4f4ddfae5f01d4dcafad2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff21e1ffd46c40baae4e482040dcff60 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1163fd05fa4321bc2fabacb90369d9 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8efc4491eba2456191fa91134d55f001 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0280ea42650f498ab6494e4c87219be9 : SUCCESS in 2m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/735961b473a248cdab854cc171272151 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3aedb35696340ffaf72f2133a8b4bb3 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c92327121544419a98c461c5dee38c14 : SUCCESS in 4m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9cc9bbaa4b0d405f9d40022a1ab88c8e : SUCCESS in 2m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/601b1b3a796046c7a6674f00c1732fcf : SUCCESS in 5m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/05199961ed5a4bff906a73cda28f2f91 : SUCCESS in 3m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f8e6f569a02846a4900c747d2e442c21 : SUCCESS in 5m 29s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a9a7a0efb2fb4f4cac3501e1ceb1602f : SUCCESS in 6m 06s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/dc8c5cb1353b4c2f987573018765b1f1 : SUCCESS in 4m 42s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/abcf01560c8545cb8274ace69feebb6c : FAILURE in 5m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3958e36a65cc45df8de9aed8410bad02 : SUCCESS in 11m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9e411e851d5c44dfbb6fd5fa7e754548 : SUCCESS in 4m 51s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0fb456a0f68c43778a15aeb80bb5ca52 : SUCCESS in 5m 04s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3c5731af8519460b950e778f98607d77 : SUCCESS in 4m 08s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/42403611c9ed4d5eae9e0f7051e5ad89 : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ba702176b6646198c0009614f5e869b : SUCCESS in 14m 49s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c211e7bcb6844b349421a7c6dc84162b : SUCCESS in 7m 31s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/659318e6dc2f4633a0bb31fe4720f82b : SUCCESS in 7m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0e80a12fe0f743cfb22790808ffd3164 : SUCCESS in 1m 38s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/62ad6a12034d4ccbaf68ecc7e1caf866 : SUCCESS in 8m 16s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8aa663e52a6642a2909ac279f61e0ea9 : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/537d9e8ea1214ce489681da3af7c66ed : SUCCESS in 2m 36s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a257bbfbf2cc4e028049c728b9f3f0aa : SUCCESS in 10m 14s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3008fc03ad804498a61790b093ad7278 : SUCCESS in 24m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d08d3e0895a8456b9804b78924a829d0 : SUCCESS in 6m 05s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e497644694c84aaeab1409a6d6df6252 : SUCCESS in 13m 08s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1ecfc40edd1a4b99b3d32f9213f6c1e3 : SUCCESS in 3m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30b39f01ff8b46768e2a748d5f4c944a : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/286b890603cd4f70a6e6e02bd524d617 : SUCCESS in 23m 51s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e8048544e367499fa4993228ef916f2b : SUCCESS in 5m 09s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7689fb218ca44587a834c7dd100e85fd : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66c422cfa9c047938bde864474f0f4d8 : SUCCESS in 6m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/cbf51876380c49b2b6851eefc7e45530 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fcf4ec0976b24b67850af22664dff118 : SUCCESS in 4m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcf123b8eb434beebb0071b4d00bceb2 : SUCCESS in 36m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e054fafea11b4e618af281e12a658398 : SUCCESS in 1h 41m 10s","accounts_in_message":[],"_revision_number":2},{"id":"6976e8b7070e11251bd471ccaa9d624759bbee81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 06:03:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"50bb70e353634a07571318f8563fd2c538ca0353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 07:52:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72b8309df31344998ab48a897a01aa5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d46034492d24024a9afd84d6724cfae : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e81b5db866044d3a06dd927e4622f21 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34c38d2ef4a74de6a8d8af7e3760a048 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95a5f4f1212743e2844b53b8bdb67280 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/87f932815a8c49729bed0f929c071cc9 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a48db37663549eebcb8864deb203ef0 : SUCCESS in 2m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d1d7bda4ce84a068a8e739a9f9ac09b : SUCCESS in 3m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9ae91b9c921144e2801c19007ca94725 : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/46350bb836de45b49c824609db259eda : SUCCESS in 3m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a3bf8a5efa6a4ade9217e1bb6954fab8 : SUCCESS in 5m 56s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/61d1dca72ab54c959d734d8b10689af0 : SUCCESS in 3m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b3a736104902425d845390e163599101 : SUCCESS in 3m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6e34232e5fe84fbb9c3c0caaf6d2d1c7 : SUCCESS in 4m 40s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/aec7c50e444c4f8e94cbcc702e4f5c09 : SUCCESS in 4m 42s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/623ec635d3f141b8970e61c519871af5 : SUCCESS in 9m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4aacefce204445f4bcb411c55f49bc99 : SUCCESS in 5m 09s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8f82360288514885a21ebcb2af546b77 : SUCCESS in 5m 01s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d1824d5376034bcdb8d0de53dd3cab77 : SUCCESS in 7m 40s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/dc1c714ec12d4e9cb561e05c8ff3d470 : SUCCESS in 7m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c7a66762d524945b68cff30722a3e33 : SUCCESS in 6m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/41fc628a365641b281ab0d6f58e4c900 : SUCCESS in 6m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d41a971b7fc3480ab28e298d80d41505 : SUCCESS in 11m 47s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f9a61abe869e4d11a1f02a3e62a3012e : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e05a63eab7604b028a8ab3a118c6aecf : SUCCESS in 9m 17s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/27573270488a4f03b182e1a885886c38 : SUCCESS in 4m 34s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/de85e07456c7401093244fd0e60b4a93 : SUCCESS in 2m 45s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/1178bdd2c1a34cffb4b5891f33460190 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/099e309fd9114484ac3c90149b2ff4a8 : SUCCESS in 31m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/24809e07ea8d4d8dbe4aca98e5f403fa : SUCCESS in 3m 48s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/32b935618a254cff85e439d3017d4b5e : SUCCESS in 10m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/af748519a0414c4abe2d97957fcdd0ad : SUCCESS in 3m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1afdb63cd629463e803a3b516833b386 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e9aa270b68a4571aa19835da788895e : SUCCESS in 23m 13s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/cff93476aa724ffe832775ac8d5ca739 : SUCCESS in 6m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/93222c1201ed4e8baa99f413ab8b375c : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2127af25249241a4a19cb24c6d450920 : SUCCESS in 7m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/37c216c10ee84e2d895378485c35adfb : SUCCESS in 17m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2458f64f06364504a6e090328d490e9c : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8b9da4bbb384ac681360b7ca82326ff : FAILURE in 1h 47m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c892d822d02438b9a90b96feba5ef91 : SUCCESS in 1h 36m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2c4736b3c406389e7b8827484a1231b56d216845","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 08:14:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba79c3591e5ee50ff9b77d43ad22aefb06401c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 09:14:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7989c12df3440a5b9cb226a63190674\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7d431c2b1a405991aeba186c04f8ab : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80446e3747f3402ba6392a7bcf248b7e : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae3bf08f737c4f3ab8c1310973d2ba7e : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6fa2b21543e4f42863dda602b7357c1 : SUCCESS in 2m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/562af7d369d54d6bb263dbba32f860ee : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36f4d27d049d41f1b285e09440729abd : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a29f7f59fedf42049eb4874124542c2c : SUCCESS in 4m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/03a298adac1f4c158227b8e26d98c281 : SUCCESS in 4m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4284212aef014655add474cf124910c0 : SUCCESS in 3m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/502b812f6a644877812e56835ce59b87 : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6f3b6f635def44f9a0b09d6c9fb9f1d3 : SUCCESS in 9m 54s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d94254ea7d3245f38f49dbff0b514c12 : SUCCESS in 4m 01s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/873bc49598be4b0489313a08c55de00d : SUCCESS in 4m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ee4be103321b43a98dce36490701a49e : SUCCESS in 3m 34s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5f42b495e8654ee9a23281c33ef53201 : SUCCESS in 11m 11s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5c602278475c4787ab8851116a767cb0 : SUCCESS in 4m 29s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2605bd4841f84b3dbd0cb4e165171806 : SUCCESS in 4m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3a8bf8c9d88a4040984fb9596bc8de60 : SUCCESS in 3m 32s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/36ade0b01ddf4d528297eb4b29b4ea45 : SUCCESS in 15m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/503b22140f374ca383978832a69ff1e4 : SUCCESS in 6m 44s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/075e2764d088476d90210e625252eb8c : SUCCESS in 6m 15s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e894765319dd4589a77290ceedf9e2a5 : SUCCESS in 9m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/3ceeb5081eda4ce5b15e3216c5270052 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3e9ee536ed8d4c1fb5de545ce680f91d : SUCCESS in 8m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/366868f24f19438d9c07df5a94f01a12 : SUCCESS in 3m 33s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/46c2e94b6e684b6f86ca4a45b49f3a18 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/387282c71e1246acad538c77d3d4a78c : SUCCESS in 9m 29s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/5cd521dd7b3e4ac8b224b26781deb5c6 : SUCCESS in 19m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d28efbf9760d4128b9f9ba8a4628de37 : SUCCESS in 6m 14s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b6d2772fa56b40dca4727642f822fe22 : SUCCESS in 12m 41s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/06b592121414402183ba15985cfe02ef : SUCCESS in 2m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51ffd5d8313848bc8ed5305e6774fa2e : SUCCESS in 1m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4f021c68ddc42038f85d725340193e5 : SUCCESS in 26m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/60f73651fdfe4281aa1ba75a42a392c6 : SUCCESS in 5m 23s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/266f36b8d3a347f0a5cd267d318ce2b2 : SUCCESS in 3m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44c0e95831a34840a92b9cb418fb6938 : SUCCESS in 6m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/5e02901a124b43e6945d6ab1baaedca6 : SUCCESS in 18m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b02ff3645319449eadf95a18f7969d94 : SUCCESS in 5m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa47d968da2a4560af3ce85c7d0e058b : SUCCESS in 47m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a1127d716394a7ba04648397d94cfe4 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":3},{"id":"6b6df342f0877b91d786da6ed4a82a4a5fda8f3b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 16:01:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ea944751f5a436d9eedbe56db59768a8fc64382","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 16:01:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c18e8193aeef2e01d4290a1313055fc5573003d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:22:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851001a864854d88b5b9c27f2269706e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d3664815714710ad6ac67ffccf1fba : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cd052c6c384374ab27f766845702ae : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/750565a2af11461b8e615c82f7468654 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2183288a1ce4cc697df7626d6c6fb6d : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a323eb7ad6364900af9e53d4db8a2242 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d933a8ceb43c48ba9556518d515f2e5e : SUCCESS in 2m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2d25126919224ce294327032be926bfd : SUCCESS in 5m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/164dd20c2aab4914a9d53317afdb0602 : SUCCESS in 9m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f49fe0624ef14e799c62b02515d28810 : SUCCESS in 8m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c505c29231f942b1b8ccf6342e0ce95b : SUCCESS in 7m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/73aeb2cc8bd142f09c1f00365e5e4a82 : SUCCESS in 9m 34s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8c2c2844cc064168b9d3ea7c32b6efa2 : SUCCESS in 5m 57s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/56594215b9044c11859e11ab90f95bcc : SUCCESS in 7m 11s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5b84595e0ca04973af4bd7ffc5a788cd : SUCCESS in 7m 02s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/fb82bbdfb36e4539824f0b30e344d5ac : SUCCESS in 16m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f78ccb78d854a4690e0a470e4e2b3c8 : SUCCESS in 6m 18s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4c575a07484c4bc09800a2d3b1d4186c : SUCCESS in 9m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/7f1c690701144f3dbbbc73b9faabbc90 : SUCCESS in 11m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e2d91bcd90db4f9599c1de8c1e2b0f12 : SUCCESS in 3m 29s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/1f380ddac74a4a7cad53994bd7fd32ee : SUCCESS in 8m 39s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c3acb272eb374d308554bc0131368e1a : SUCCESS in 29m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec0684f5b1a14e608c5cb23c60a314d1 : SUCCESS in 7m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1e5f0ec743594295ae681fdc0dfda7e5 : SUCCESS in 13m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/62c75bc51ad34aaa8d63679c305aa1d2 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3063fc4bbf54808a28ca2cd0c766935 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9f72786c91c4c349ab1d0e49a18133d : SUCCESS in 25m 23s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/56f6a8b688c5427ca2a57a472f4afb72 : SUCCESS in 5m 41s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/630cf021796a4ece9bb69357f603c6d8 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0493f7045a84449a0f6a0f08db16fd6 : SUCCESS in 9m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/59e6b83ef4e7458a80874dbe036dc50e : SUCCESS in 15m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/45c403c665d8404ca6fad1c741440f64 : SUCCESS in 7m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd0c780030a34fea925ecb4651c38026 : SUCCESS in 54m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f931fc88e924829aaa8c5f0ae3e4dd8 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4f0a36c3af20312d1e56347fde4ef801c2de7669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:22:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0c1b5e77f9aada1ce55ea5c60f6d265b1d6c199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:23:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9200037a0fa4e0d83db034cd01ff62b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab8d0d526ca4224a6ab28050721dbc7 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6872dbbef8ae59aedc9696a33dcca3ab6538e00c","revisions":{"18c597942c341b54552538f37ceb3e0800195e29":{"kind":"REWORK","_number":1,"created":"2026-04-25 08:28:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/986249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/986249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/986249/1"}}},"commit":{"parents":[{"commit":"f2400d89cc9eed3c682d21cdde96de84c7aeb2a0","subject":"Merge \"Fix errors in detail.py helper script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2400d89cc9eed3c682d21cdde96de84c7aeb2a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-25 08:28:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-25 08:28:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib909ecfebb541e306444f4ed6483f5d48dbe807b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18c597942c341b54552538f37ceb3e0800195e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18c597942c341b54552538f37ceb3e0800195e29"}]},"branch":"refs/heads/master"},"c7e32f45d70f721fd8669449c928bdde56f46059":{"kind":"REWORK","_number":2,"created":"2026-04-26 09:44:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/49/986249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/986249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/986249/2"}}},"commit":{"parents":[{"commit":"f2400d89cc9eed3c682d21cdde96de84c7aeb2a0","subject":"Merge \"Fix errors in detail.py helper script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2400d89cc9eed3c682d21cdde96de84c7aeb2a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-25 08:28:20.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-26 09:43:22.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ib909ecfebb541e306444f4ed6483f5d48dbe807b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7e32f45d70f721fd8669449c928bdde56f46059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7e32f45d70f721fd8669449c928bdde56f46059"}]},"branch":"refs/heads/master"},"6872dbbef8ae59aedc9696a33dcca3ab6538e00c":{"kind":"REWORK","_number":3,"created":"2026-04-27 06:03:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/49/986249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/986249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/986249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/986249/3"}}},"commit":{"parents":[{"commit":"f2400d89cc9eed3c682d21cdde96de84c7aeb2a0","subject":"Merge \"Fix errors in detail.py helper script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2400d89cc9eed3c682d21cdde96de84c7aeb2a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-25 08:28:20.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-27 06:02:39.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ib909ecfebb541e306444f4ed6483f5d48dbe807b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6872dbbef8ae59aedc9696a33dcca3ab6538e00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6872dbbef8ae59aedc9696a33dcca3ab6538e00c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
