)]}'
{"id":"openstack%2Frequirements~986346","triplet_id":"openstack%2Frequirements~master~I23690dbec6cfbcb8a3a013fe5ebcecbb0a1c473d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I23690dbec6cfbcb8a3a013fe5ebcecbb0a1c473d","subject":"update constraint for openstack-doc-tools to new release 4.0.3","status":"MERGED","created":"2026-04-27 10:45:44.000000000","updated":"2026-04-27 19:20:33.000000000","submitted":"2026-04-27 19:18:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986346-new-release","meta_rev_id":"c33a299d226d2464384feb580d5aba2b0d1b2822","_number":986346,"virtual_id_number":986346,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 19:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-27 17:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-27 17:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 12:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 17:20:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0983b3643ea19713ee8036ddd65d96399fd73fc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-27 10:45:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3265dd29a50e5dffa54340142d2ac99286f4aa16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 12:17:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cb8f6d7ee25438083ef3148ea74e7d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eaeb81cd00c4b538aa80360bad6b163 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8806bd41ab9f466782c435112fa17972 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0b98e124a0f47a89d9c9cab83ff174a : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d033036f955845089fc8e7445a0ab6e4 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bce2b137784048a4aabebef2ec5ef826 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b93d543004f24865a135739d6a518819 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c936cff3f60542b49e34d1b8bf3da063 : SUCCESS in 2m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bd53b7fe0eae4e2f8d6b1a37bd5735ca : SUCCESS in 5m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0d814db901004e83afcb01086e748367 : SUCCESS in 6m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e51f6119625b43c1af4773cb45e663ec : SUCCESS in 5m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/58d7d61a04d24fbb8ff2ff7cacc8337c : SUCCESS in 9m 33s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2c0803cafb85422e87552788337ac8f2 : SUCCESS in 3m 44s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f36c05c8a41c420d9642ff40b633a164 : SUCCESS in 5m 35s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/fc8a7f69eb7b47d99a821463b6b47d49 : SUCCESS in 5m 02s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6b87e05474d94ec1915062571ae55e1d : SUCCESS in 11m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/45e493b2cfbf4206b60fcb1171148e48 : SUCCESS in 3m 18s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/da0bca6bf5d9401b8029584539fc1a32 : SUCCESS in 5m 05s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/370c2d06d946419eba5b0c09a036b25b : SUCCESS in 6m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/056e0835136541bf8e98f0f1b3730fb6 : SUCCESS in 7m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f339b3a538b4dd9b4aa38bb7822813c : SUCCESS in 8m 07s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/24aecd7ed61842f097fa16299e1d62f4 : SUCCESS in 9m 19s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/437ef5fe146b4ae68bcca704bd09193e : SUCCESS in 9m 52s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/91de6ffede254828a6217300bc7d02ad : SUCCESS in 2m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2cb709b1e87b4f6a95b4ac221f8c16ce : SUCCESS in 5m 00s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/1a5882cdf27d475c85496989945ccf97 : SUCCESS in 6m 21s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/82a497d512a341d8962194039fa92ae4 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/543e2cfc04f84a96b817e71b0448c318 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6dbca53407164e508002b4df6be63ecd : SUCCESS in 16m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01c50115219a40bb9765a1024b4fb1b2 : SUCCESS in 3m 41s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b491916f3cf44f66b22e04e037367a67 : SUCCESS in 14m 59s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c03838351747463296097ed6663e6432 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/712ef646301e4d658dfe2b920edb8d86 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1e19bef874a4d8a872e3119467f9459 : SUCCESS in 26m 24s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/87116d9e34554a67bd7469f0f024a37f : SUCCESS in 5m 19s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5ea77663e3064b8c92b71341254aa242 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6c24d1b4d4643c3aa904b9abefb0ff3 : SUCCESS in 4m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/74cec356a20f489395a278c442af7fe7 : SUCCESS in 17m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ebcb8e09d40413da712874d77df700c : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99992d22c8f4405e8c2ac86373fba17b : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f36fadd02f043ff935f29c335fa68ee : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7336bd0a960dd840fb233371e0e49676f3e84278","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 17:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f4d5e14ba5965c0e7e6c4a63b3d8377e701f862","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:21:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70b20dd3e2de26b47831c348c6813d72a2d8edd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:18:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7519e8dda454e63b1d92123e797e623\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c8e64aa2094466aef305b21d9f9791 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be045468468c47ff910335305b73ef86 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30c4fc6db4dd4307813487d60c3d0f84 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86b18d4b20984860aa4d0000a06edcbb : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4383108c932d4fae9641063c348c22b1 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe57911e977e47c69bdb29e1fb98c047 : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c5b38dfae704a89af8ecd0c9f7fa47e : SUCCESS in 4m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a833f57580cb48dba2507ba8d6ac1245 : SUCCESS in 5m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dcebcc725680440599f972b39a9e3216 : SUCCESS in 6m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7c580824ba204ccab570ddde9d3806f0 : SUCCESS in 6m 38s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/2673f906f88943989b4cea3f1f0fec1d : SUCCESS in 11m 14s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d91d5ab8fb624607a76c3a2e8bbf51f8 : SUCCESS in 6m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/17467229a708424f8bd44c45b47eb306 : SUCCESS in 3m 53s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a059d4cd33bf4d4695b8000d11501bdc : SUCCESS in 8m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b8d297b44a674e4aba206f774cffaf87 : SUCCESS in 15m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69d3e4b04f204e06b4a2db3e49403c5f : SUCCESS in 5m 14s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7e6ceafdf56d4158b6139fa8ff9ff37d : SUCCESS in 7m 22s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dcfa804818fb4b1eb75a158512a75b2e : SUCCESS in 11m 58s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5ea4e636bfd24e168549ffd41ca6aecd : SUCCESS in 3m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/823406bfca47499484a9869ca349ba56 : SUCCESS in 7m 48s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6ae2178863a144cf8ddf343152265664 : SUCCESS in 18m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fb03453b49734c2b86a470e20a7bc1e6 : SUCCESS in 7m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0d993f49337a458c909d0ccf79917f9f : SUCCESS in 9m 09s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cd81df82d4c44366be77e6c7bdba7e18 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a1cffe0e8c8492bbe21ee0bba3beb4f : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd8e8fc68de24796bd0752e40db8c957 : SUCCESS in 29m 39s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/aca78fbb0f5e4292a3ad414251a72cc2 : SUCCESS in 6m 06s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/d124fb8379694199b2f553759c74195e : SUCCESS in 3m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0421933ee2124a1184755a4176fa93fe : SUCCESS in 4m 59s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a0273b83ab874de08969662101d657f3 : SUCCESS in 14m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/362530a85ef94ef6b15a1c8806000b5e : SUCCESS in 6m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56d2f7280f9d484086881aabe79f170b : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9bbaf29cde54eaf916323292300f39e : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c343f3d6f8c75d9b3fc8bdde60a5d9cb09273be9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:18:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c33a299d226d2464384feb580d5aba2b0d1b2822","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:20:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147ed09922e14373ac090d6304ff9ff1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9538772d935f41aa9368214a06670f7f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bb240031ea259f44473d1f6e9db4fec00943aab","revisions":{"3bb240031ea259f44473d1f6e9db4fec00943aab":{"kind":"REWORK","_number":1,"created":"2026-04-27 10:45:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/986346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/986346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/986346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/986346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/986346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/986346/1"}}},"commit":{"parents":[{"commit":"f2400d89cc9eed3c682d21cdde96de84c7aeb2a0","subject":"Merge \"Fix errors in detail.py helper script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2400d89cc9eed3c682d21cdde96de84c7aeb2a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-27 10:35:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-27 10:35:43.000000000","tz":0},"subject":"update constraint for openstack-doc-tools to new release 4.0.3","message":"update constraint for openstack-doc-tools to new release 4.0.3\n\nmeta: version: 4.0.3\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I5bfef8ee3fc438991b555e4f45822aaeb85c5645\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23690dbec6cfbcb8a3a013fe5ebcecbb0a1c473d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb240031ea259f44473d1f6e9db4fec00943aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb240031ea259f44473d1f6e9db4fec00943aab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
