)]}'
{"id":"openstack%2Frequirements~986355","triplet_id":"openstack%2Frequirements~master~Id42446ff395639e62afdde5f17006ec5610105a9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id42446ff395639e62afdde5f17006ec5610105a9","subject":"update constraint for openstacksdk to new release 4.12.0","status":"MERGED","created":"2026-04-27 11:27:52.000000000","updated":"2026-04-27 19:19:49.000000000","submitted":"2026-04-27 19:18:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986355-new-release","meta_rev_id":"2ea370ef7a133051ca3832efb8e669c42f2a51d8","_number":986355,"virtual_id_number":986355,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 19:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-27 17:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-27 17:19:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 13:25:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 17:19:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8d13c376b48c24e647ac68cd53c0e11e18c5466d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-27 11:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faaabeaf357a5d6004407f0fc27923e52a8207fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:25:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb3fe9fbd8345fbb4bcad2f62f1a4e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8487fd22b3ce4ac99e4fb03333aa7951 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9341c5c030754f98beec2e69394decc1 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/044b27aa39b041da886d46627695c3e8 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bad05d3cba2a4ea48d642561c133e83c : SUCCESS in 5m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e552cd37ed8f401898437560f2cf3302 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebf5078992b743dcbbd5946090bee55c : SUCCESS in 2m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54500f73b36549d497926b2469ae99d0 : SUCCESS in 3m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e1f6bbe6735a4164abfb3ff8c24ec926 : SUCCESS in 5m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6e0cc7b8913c47c6a1e5a2e63782827a : SUCCESS in 6m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9d592a9b3d094ac4a2f8850ed04793c5 : SUCCESS in 10m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8b08253a04834ba98043f85200abc2d6 : SUCCESS in 8m 15s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c7905c3d2ab646aebca84eabc9d73544 : SUCCESS in 3m 14s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d200f415d3474b588254775a9995c285 : SUCCESS in 5m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/dbe30097e0974c57b9ff3bca6b41cd2a : SUCCESS in 4m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/662c35676c3e480f8708a0b37a21cb08 : SUCCESS in 10m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fcac7673b81548828c0992b511971886 : SUCCESS in 5m 31s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/22d882200ea24eafbcb40434f2db76b1 : SUCCESS in 5m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/16592e7a04b14d16a02450c290980267 : SUCCESS in 6m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/dd4ee4fbbde14baba382f3911746c87f : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdbedfb0d08246829e5f974f7c1b1d67 : SUCCESS in 5m 24s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b9ca3c81a8fc40c5a9bec25800509a67 : SUCCESS in 5m 57s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4737d2d9cf2e4a39bbb9de03bf20af57 : SUCCESS in 11m 24s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1a95f8d263a94891804f018d0a496a4a : SUCCESS in 2m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e9e77240d2564617a2f3aaebf0a59d10 : SUCCESS in 8m 47s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/20351175f10949c9abb403306c735539 : SUCCESS in 3m 35s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/a4eb70f33593418a857f156e9e73b992 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a141ba1d3c0e4bf3b6598ae8ef4d25e4 : SUCCESS in 10m 17s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c14a9819b96344b39a1ea0d22990d3fc : SUCCESS in 23m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd51d3836fed45da9bebe939058bbb0f : SUCCESS in 8m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f43e1de310c94b8a82ddd3ba4b4d5923 : SUCCESS in 12m 52s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8f36fdc5273448e8a7041e40bd5efe91 : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6db177859db4f37919d3717124d6d9b : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f57daf4e9c54641a85a1343002ab6df : SUCCESS in 25m 03s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/258a2304bb274c8cb34928912ef14e89 : SUCCESS in 4m 36s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e5b42ba7fdac45979a7d137796827675 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fa4a77e25c741708336d91bd01230e8 : SUCCESS in 6m 42s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/09fd8ed096834f99b8dc5ab8d92be181 : SUCCESS in 18m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0ea007d390f04bce97f4d6fa0ea9c20e : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe1ae3bddfcc45d0b7789599985d5734 : SUCCESS in 1h 07m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f530fc0e42d48d2a100900ef398ebec : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2e614d07cc48beff981c72ed0efada104fa6318a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 17:19:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d14fb59d3ba4a38ec871db76f56ba0e8b3834388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 17:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e614797b74aeec705ccfbb79a726e2c0fc9d0939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:18:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29583f0da5194e89a2f430bf256ad803\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c85248db3a4fd5b2c48f6079e80e8b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebaec8ed7ef349c49c03e9f5ced1e41f : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a261d7efa08c4eacb6e863135c9b08de : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1d3d4bd878e4090b8970d1a51f46ed8 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6da47a979f2e406d8475c3801cd91fa6 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cfbd201ab2204a2fb29ea9b1c1a355a6 : SUCCESS in 4m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9891e3ad77545029590daa8e07e7055 : SUCCESS in 5m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a213a33775054f0f8237b82285a39887 : SUCCESS in 4m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6f62f82ad25f4373bb54ecb2422f713e : SUCCESS in 4m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a12ac392e7d94e27875ec9c87848679f : SUCCESS in 7m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e834da4ada264231b3b3a7c85f5741f5 : SUCCESS in 12m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/99e1618dd5274ad497211c4f3742bca0 : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5def8b9629a44dde86ca78f77c13a22c : SUCCESS in 5m 26s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e1c7301531e548e7b8f07dfa84b799e4 : SUCCESS in 6m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8e52a3ca53624783aaa83baf855c3f8b : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4336dfeb21c48c1b28402b5235be78b : SUCCESS in 7m 27s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1a7a0c3f5357420cad1ef35339941def : SUCCESS in 6m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e150c91a22f94b31964f72b2b4ef592c : SUCCESS in 11m 51s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/eda64ccfcc7f4322bfa29f10cc991d22 : SUCCESS in 3m 00s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/725d0c4e37374226b50296e5c5525d07 : SUCCESS in 8m 56s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3f4b2794ce2346a8a4f42a1e46775a13 : SUCCESS in 26m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e3d82beb4f6942ffb6536578150fe4d3 : SUCCESS in 7m 47s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/99c98d7be5ce46fab6622b0f591da2be : SUCCESS in 11m 27s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/76686780ae7a4c32876615e8929a794d : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a30457fa611742bc9163737e88a63379 : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2beda72462dd4952b5f4c12eb4f7ada5 : SUCCESS in 26m 19s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/719b87c6433848b1b3d41b1346fc575b : SUCCESS in 7m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6a5b1cf8b5184c55a24ce94023d8fc24 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/20c63fb8ad484e3e920ad88851e877f1 : SUCCESS in 7m 16s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7b03c99de3fc45699abb3367205a40ae : SUCCESS in 19m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bfb311a6858444dea5947fd7c8ba682d : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69cf15a0e44c4ce7b1bc247fb1c2ad2d : SUCCESS in 1h 09m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae98962cc9554b8dae1904d709b69bbd : SUCCESS in 1h 52m 01s","accounts_in_message":[],"_revision_number":1},{"id":"94ebe47061dbd9359bf585a1a06b2a33cdf78376","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:18:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ea370ef7a133051ca3832efb8e669c42f2a51d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:19:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b051b6c359476ab4c72407308a0ceb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7781a38b402b4c7c9d2022012416cca6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f93235e6781cfc9e24b0e0c45d89ba6e36c46db","revisions":{"2f93235e6781cfc9e24b0e0c45d89ba6e36c46db":{"kind":"REWORK","_number":1,"created":"2026-04-27 11:27:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/986355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/986355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/986355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/986355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/986355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/986355/1"}}},"commit":{"parents":[{"commit":"f2400d89cc9eed3c682d21cdde96de84c7aeb2a0","subject":"Merge \"Fix errors in detail.py helper script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2400d89cc9eed3c682d21cdde96de84c7aeb2a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-27 11:17:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-27 11:17:51.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.12.0","message":"update constraint for openstacksdk to new release 4.12.0\n\nmeta: version: 4.12.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dylan McCulloch \u003cdylan.mcculloch@gmail.com\u003e\nmeta: release:Commit: Dylan McCulloch \u003cdylan.mcculloch@gmail.com\u003e\nmeta: release:Change-Id: Ied48263d49958464e6e56c8558216ddea642b20d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id42446ff395639e62afdde5f17006ec5610105a9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f93235e6781cfc9e24b0e0c45d89ba6e36c46db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f93235e6781cfc9e24b0e0c45d89ba6e36c46db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
