)]}'
{"id":"openstack%2Frequirements~986803","triplet_id":"openstack%2Frequirements~master~I840ff1d60a4ae3f0b0d3bd132687e27784433b01","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-30 19:01:11.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-30 20:19:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I840ff1d60a4ae3f0b0d3bd132687e27784433b01","subject":"update constraint for oslo.privsep to new release 3.11.0","status":"MERGED","created":"2026-04-30 13:33:58.000000000","updated":"2026-04-30 20:20:40.000000000","submitted":"2026-04-30 20:19:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986803-new-release","meta_rev_id":"17027a243a9a7b1031a9aa39b1b165c45035626c","_number":986803,"virtual_id_number":986803,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 20:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-04-30 19:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-04-30 16:54:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-30 19:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 15:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 16:54:32.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-04-30 16:54:58.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-04-30 16:54:58.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"418765902efecd0d0ed3bdc0719861d5d6a64202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-30 13:33:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ecfc7d279327fed70b02d2d1f5975d8f67a554a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 15:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef9a2d3d4a54bcba445560466797bca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d0c0d63bf04a8ebc913be673eaa2e5 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2fc8e9ccb84061b04f8c04d66d7670 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28fdf78abe774700b8a72ebe2ca9ea82 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a39b5930e1544e7a98eb79566a3b3ca6 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb2ba9fdb98e434287cbc871437bdcb4 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8320c7869e94dc1b94ddf9c18c9a568 : SUCCESS in 2m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fabfe977e658410cb7a0119cfcceba80 : SUCCESS in 3m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e8348141256b47079c605ceb3d5864d5 : SUCCESS in 5m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0488951344f24593848265b45a7f17a0 : SUCCESS in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a67b09a1e4174153aba404a96e3e1239 : SUCCESS in 6m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/74bb826ef44f41da861c0fd7b29d0670 : SUCCESS in 5m 54s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/91a36369536f42e7bdf26c43aed8eacf : SUCCESS in 5m 54s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/eb1a0e5222ca466e9639ca060ee479c3 : SUCCESS in 4m 54s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/711e757935a8426fbeae4fbc6050f0ee : SUCCESS in 3m 29s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8b9a75f596d548d5969bdd2b0b29d45c : SUCCESS in 10m 51s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/80ae203c82ff47908e8b28fd0ffb60e8 : SUCCESS in 3m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ad3da9c61d4543e3ab5d19b4b7aca251 : SUCCESS in 4m 51s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/09f793fb87cb45aaa9b3912071f2806c : SUCCESS in 6m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b1b636026e6d4d25a055eb1c22252af4 : SUCCESS in 8m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dae58727ecd14a49bbb5bc31e9947d56 : SUCCESS in 5m 36s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2fd2854292124aff992e45269aa219a8 : SUCCESS in 4m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/297e8a1007a247f6800a9bd8eadc7470 : SUCCESS in 11m 25s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7a41a256c2b44732a4b1529aeec1581b : SUCCESS in 2m 08s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/50c86ae443d843449af7f21a3dc299d7 : SUCCESS in 7m 42s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9c4ca75ed9ff4d05bd45dfa1187decac : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c7cb921c7bd349cf98a3d1f532043047 : SUCCESS in 4m 12s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/60174008ffa846d69fa4f756ef53a101 : SUCCESS in 10m 27s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fef19327e0804390af6169fa901cbce1 : SUCCESS in 27m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a7b280cd8f5047a799d3a7eed4aa1325 : SUCCESS in 7m 04s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0808b4461c8b437c82136f13b2861168 : SUCCESS in 13m 24s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/9fd1d658e67e4a76961e77a771167a89 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e825bb19324476a97d9b9858c14ce5c : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f3633fc44ae4c6cb7fe462ebf0d128c : SUCCESS in 25m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4a21d6f12f7e48368701eaec21478ab1 : SUCCESS in 6m 24s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/786d71db33e8499b8e97a7283d1f4524 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/335b345e7e194933b1a3d5775e631d1f : SUCCESS in 7m 00s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f5585e38bb0448c0b48ba55f4d2d519f : SUCCESS in 19m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0d312646627d4878beef39e0c178fbc5 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5051f50b566d4d8bbdbaae3fbb7173cc : SUCCESS in 1h 06m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cfcce130ba344849a4f8af1e3c8537a : SUCCESS in 2h 06m 11s","accounts_in_message":[],"_revision_number":1},{"id":"687aca63ee6e264dd680c25e823a5ab133cf7606","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-04-30 16:54:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f248fc7e0e52745cc897e2981530c5ac4d3cb90e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 19:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de3efaf19453d6df939f43f1d258c80047b34e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 19:01:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4187b6e74a8c48049b3e5969d455bff3c58beead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 20:19:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1235bb0c60ac436499c988e7c6df52d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6fa5ea554343d196119a466b0c37cc : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf31a8cb78d640eca0421f662e165854 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e484482923b0442db18759735c1b819d : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02999992c3224ee4991414ca588174ec : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9440665efbb44950b67cdc4737991f38 : SUCCESS in 2m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96bc9255b1fd4ca28a85df72ae47ac6c : SUCCESS in 3m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/246d492d19c4403aa133cf176bdd8ff0 : SUCCESS in 5m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4c3a93081b6347299f718c93fc367a63 : SUCCESS in 5m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/33b6e6211de04841949a686d5db74bd7 : SUCCESS in 6m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f27a213d8acf498bb4ae1dbfc47748e4 : SUCCESS in 6m 08s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/33bc4752e2834b1db5aeb7ed81bcfb47 : SUCCESS in 10m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9c0c665a1ba9492e8bd6ae4f735cb00d : SUCCESS in 5m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/df24a3b2a0164401bbdb1980e4c5064b : SUCCESS in 6m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/614ab4d3e572480e80d8b858e4d84c96 : SUCCESS in 6m 45s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/37716eb6aaec49d3889837dfee140bcd : SUCCESS in 15m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/374cf96df1a04b688232e4fc981db11c : SUCCESS in 6m 48s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/61d2cbfd4cce40feb19ba064df23ea33 : SUCCESS in 9m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/001d4589abbc494f8d9e1132aa921145 : SUCCESS in 11m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6da56650053f45638ad5593d369fea83 : SUCCESS in 2m 28s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/912779c58439491489077629211e4a5d : SUCCESS in 9m 01s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4398256fd64b43fdafd0fac96826060f : SUCCESS in 24m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4250572580db434dbff0bef8290c8120 : SUCCESS in 4m 38s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/07d26107d8064abab61506f99b52300a : SUCCESS in 8m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/56d959037ed0452f82e201c3062a5312 : SUCCESS in 2m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58109cd29b7e4f7e969bc4900cabb64d : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e76b0ed372e46d88f279ee4eea58bb3 : SUCCESS in 25m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4d4f301a1c4b4573abe2565d01a26f42 : SUCCESS in 5m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/914730ea770942a59fe15373721ab636 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d76755176294ec98f29045602afc34f : SUCCESS in 6m 03s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9d5984668f624b2aaf3493404650d051 : SUCCESS in 18m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c0900636687f4f8f9ef0438a64f17370 : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/099d38150b3d4855aa676776ecb9c399 : SUCCESS in 1h 06m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/166ce07a9c034c9784238dbcdc093873 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3900b52c6cdcdaac9ca8c82af933d7c6dc9ed277","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 20:19:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"17027a243a9a7b1031a9aa39b1b165c45035626c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 20:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6aabfe7751442092166a13a0914ff1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916aae7c0228407f94fbc2d4dcd3b374 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed679515f6e053d20b05ffcc30b3172de864f7f1","revisions":{"ed679515f6e053d20b05ffcc30b3172de864f7f1":{"kind":"REWORK","_number":1,"created":"2026-04-30 13:33:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/986803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/986803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/986803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/986803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/986803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/986803/1"}}},"commit":{"parents":[{"commit":"2699f4c69c14b3757f98bc8a6df169fcdb52b4a7","subject":"Merge \"update constraint for openstack-doc-tools to new release 4.0.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2699f4c69c14b3757f98bc8a6df169fcdb52b4a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-30 13:23:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-30 13:23:56.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.11.0","message":"update constraint for oslo.privsep to new release 3.11.0\n\nmeta: version: 3.11.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Change-Id: Ida6788b25b6acc6622a51883d824733a1008d421\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I840ff1d60a4ae3f0b0d3bd132687e27784433b01\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
