)]}'
{"id":"openstack%2Frequirements~986968","triplet_id":"openstack%2Frequirements~master~I26073c75f5fba0c869854ce02406e2673f3ca881","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-05-07 23:16:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26073c75f5fba0c869854ce02406e2673f3ca881","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-05-02 08:18:13.000000000","updated":"2026-05-07 23:17:23.000000000","submitted":"2026-05-07 23:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986968-openstack/requirements/constraints","meta_rev_id":"5990f0d8836ed8a2fc1aa500dd4df1283af40145","_number":986968,"virtual_id_number":986968,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 23:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-07 12:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-07 19:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-07 12:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 10:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 15:48:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-05-07 12:01:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-07 19:50:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"69b6aebca9f2ce273b34d2a5cc5b019202147a1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-02 08:18:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd286e2705787970de700d475620edc66ac4520c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 10:00:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c938671e1856473fae70f8fd2c40437b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a39192882f54aae8cf5755eed860cfd : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5374678a97df4f53b8df64bfdda41fda : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/220ba12ac01b42b3a48d521645d8e577 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e9afc12ea694d27b40630f722792404 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a07bce0c6d94ca782c0eace71011a33 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/765d62190c8545eeb4f0e4bc746ae9ea : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be816b1511504e3383d9e5a0437aebbd : SUCCESS in 2m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/967bbd125ba7455a8242d1af17309f27 : SUCCESS in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cc74b0ad56ab4f6b87fd637459abe95a : FAILURE in 4m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/735afdbcc83a41089f8c5dc7f8df2bd5 : FAILURE in 4m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/23eabe53456c4bbca98bda8b995a2556 : FAILURE in 5m 47s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ab6e18f6e11640b79ae3b7c6b830bb4d : SUCCESS in 3m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9ef5fb29fe5f4dd88ef120e4ecc59a08 : SUCCESS in 4m 51s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/77fbcac33c134565baa57a2eadbaf2cb : FAILURE in 7m 17s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/aef286023bf0469eb904b1c444e3be5f : SUCCESS in 10m 47s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6681925bebbc452cbe0fab7404299900 : SUCCESS in 4m 31s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8597ee72f8f14f79bf6c565ddd4b4c3e : SUCCESS in 6m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0b4cf5ebd2694a6abbf387733e0ddd1b : SUCCESS in 8m 01s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/86038373f7dd479ea3e37aa7ee679316 : SUCCESS in 15m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3af16e23719942d299601e204be0ef4d : SUCCESS in 8m 29s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2894cc768c4c403486587f8c33f9a8e5 : SUCCESS in 8m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0d812b6848db445aaca95c4a5ce0949f : SUCCESS in 6m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ba7381635cf24f80bc655aa9c412ea31 : SUCCESS in 2m 37s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5f407723fa4d4689a71d9283293a8df0 : SUCCESS in 7m 04s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c70cd96b50a54a2eac66cf8b7ba63a1a : SUCCESS in 7m 01s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/95ffe52153434b9b8271aaa954890677 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/1aa21fa75d424652ac0c02357428a2ad : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/052de507968d4642acb57825d4e1c0ca : SUCCESS in 24m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ce19e5a7be7a492d8d329c5b3da67107 : SUCCESS in 6m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/3155488003f44d04a5daeffac35c03de : SUCCESS in 14m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cff3770b0ba44b50ada67da7a3d69021 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c71b442b17314064a21e4ba8b978d47a : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/334c262266fd4cf6a1b2a8d191e21629 : SUCCESS in 24m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/3dcb9b8e979f4c2d984cafc4c05fafc1 : SUCCESS in 4m 52s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/89af69985acc4577906da2029c893b04 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b1d4a5c68584e3981cf9e11cd2bbb12 : SUCCESS in 6m 49s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e2c3b827f80a4170bbb03034bc9cb7ff : SUCCESS in 12m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/455fecbe4c464cb5b263548129ed93da : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c278a37cda24cd3bb92b08f89c86ed6 : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8aa3a1630de4535a721d402bb410ff0 : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":1},{"id":"aa2c2c5f399efffe6d7f9d32787816525becdf0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 12:22:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5b7acd6a9c565ae6c21739a86acf8ed81f81069e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 14:07:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba14ac564d104f82ae871c334cf619fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35f83beab094d32be04654b6bbf1260 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5890b5f5af7449368af33af4629de61b : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5caa44db38ab4cceb60d4ba5260a95eb : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b468cdeedecf480b8615be0970fba8de : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ff7de1b524e49528a9c64984f7cd1fb : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/368c0b5261244f5b90f73d35266d0bd4 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11b33ff27d36433cbdf0bf3178e38196 : SUCCESS in 2m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0a74ac0aa534726bdbdd5ef458d6825 : SUCCESS in 3m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bf200627b70644aca52475376d458bfc : SUCCESS in 4m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/61054c3ffbbf4f068a18656bc7a99a74 : SUCCESS in 5m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/99ccc47a086e46c4bcd71789e2a22334 : SUCCESS in 3m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/3cefd77d59ac4ab9bed96f0beb06bd9b : SUCCESS in 3m 07s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/90c6d10547e84afc8b2763485d6bc1ff : SUCCESS in 7m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/bcad2629a5494ba6afb888f2ccdd8df6 : SUCCESS in 4m 16s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f8da89ae0a384a07993777c0287af359 : SUCCESS in 7m 25s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4d859c1aaf1949f98f4fae302750fdf2 : SUCCESS in 6m 57s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/fd55f63ca09a48e396f6b4814440612d : SUCCESS in 4m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e1ed10cb72ca476a8446df5d8b0e42ea : SUCCESS in 5m 20s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a0b991a4dd204bac8b16cc0094049edc : SUCCESS in 13m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cbfadbbfce24ab5995b1d4f222f41a7 : SUCCESS in 4m 09s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6c8058d760004b4087dd13aa2a93830e : SUCCESS in 8m 08s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dec4dce1ebf94b84bca36a3b99ef7032 : SUCCESS in 9m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b712715d908842048b623cf74b664fb6 : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/84f5e98688c74976ae6922c3b8fc0205 : SUCCESS in 8m 32s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f90a98f481714868b58d6702fe959595 : SUCCESS in 2m 23s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/654c74e2b02b4c25a787496061f5a82a : SUCCESS in 6m 31s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/59047f5f1c324b5aa4de4a2b8ac6f32e : SUCCESS in 10m 46s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/29c92166dab74a23b2fd600e0e51ec35 : SUCCESS in 19m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e56ad57f5b384ded8164f4d82be3d8e1 : SUCCESS in 6m 16s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/27ed0387223943048a62b3a7e396fb43 : SUCCESS in 10m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cd3879019a21468d849ac7b8f918a6ec : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8bd43ba644f46e1abbdfb0951ecb545 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6704fa45ea6649e1bf2411c46b4f5132 : SUCCESS in 28m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d36a672f3ac64328bf37686bafc66c9b : SUCCESS in 4m 35s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dc4b96cd8e9e4e79a423ec9e194eb9f7 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45272c0069a547df96ed11ad33c26dfb : SUCCESS in 4m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8afd1fa0bd7048479e86c84515e89358 : SUCCESS in 8m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6016f3eda135450fb9604db89b6d24ca : SUCCESS in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16c1db2b04a5466d80779ac257ac103f : SUCCESS in 59m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4ac74ffacaa487588448d0e308cb72d : SUCCESS in 1h 39m 16s","accounts_in_message":[],"_revision_number":2},{"id":"910ecf8b2b0820f13a9a0d043852a152f1fe5e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 16:46:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a20a7fdb205409a7ef52091522ad01fba574372f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:31:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/148b8b55f9ae49b19d375a63c7e3a5c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a84b2d573ed4f37bc88a7d71bd78f7e : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11590a1a6c14c88862b808a54d91c90 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f541e537c7544c05800ed91b7382bc32 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e2d963059f441a181b86ff8b7b0b1dc : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/163f70ec40bf472fa81f66592a969b8e : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c15c1f5c3def49f5833cd47c09fcbea8 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a74656919006437f86e36819ce9eb80d : SUCCESS in 4m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/84c1f3b2067245dc8153368950444e5f : SUCCESS in 5m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6d57a12bc1934e33a3472fcfd47f36c3 : SUCCESS in 5m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1fe5a8a53cb84e4fb61dc0079b70332c : SUCCESS in 6m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8f2438acc3c745c591ee4a828b66760d : SUCCESS in 5m 43s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/8f106e5dac4446abb1b2cc495ca02095 : SUCCESS in 4m 36s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/143f4f02bb464ef4ad6d6aa45e013771 : SUCCESS in 5m 26s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1bb7f9eb587c444d8af6ded4a2580802 : SUCCESS in 5m 07s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a4b5f74b2c714de89787d6e48bc6ccbc : TIMED_OUT in 41m 25s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2ad0f01e3e154dd6912f163841d872e2 : SUCCESS in 5m 57s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/dfed7a2514e6439ea8580d74b87671b4 : SUCCESS in 3m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9bc959a92c2a4cf7a28368e9b84b9fe7 : SUCCESS in 6m 09s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/fe7d32aaca1b49dab29fc36608faefb3 : SUCCESS in 15m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1db0bfbfbb1d4fa19578f088a93a0927 : SUCCESS in 6m 01s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1feb1c633af84aaba03c03d19615e2e6 : SUCCESS in 4m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0f57270e358c42d2902cc6976ee4b4b4 : SUCCESS in 12m 50s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/4689842733184a3c80261ad6bdc7fc77 : SUCCESS in 2m 39s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ebdafab49fd84be8992c9602c01534f2 : SUCCESS in 7m 24s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/2c5bc8b57d9c4c368a045195eb801086 : SUCCESS in 5m 13s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/98d67be7f61f4cc9a19fe87ff7f07315 : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/c71ce54c45c44aa6baa5a4ffd003f6f4 : SUCCESS in 9m 09s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/acb98e74cc4f4917a4b3c394d3bab06f : SUCCESS in 24m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d43f7d05de414cc39d67e30867b0a8a5 : SUCCESS in 4m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7e711440f5464a0095bde3f6191ab783 : SUCCESS in 13m 27s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/889e22a9e2a84e4a99726891eeaadb3a : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/851b269afddc4750a9202d6fc283356f : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38de09b7650849238b83a5c14a868ac2 : SUCCESS in 21m 08s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/02ca34386d544b4cb1e430b73e237670 : SUCCESS in 8m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/27645099d7dd41f88019fdef1cdfdb36 : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70fe4e3ab30242f3ac7fdaf78a2c5dd9 : SUCCESS in 9m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/760961623ed64ebf8d3d861b639cab08 : SUCCESS in 18m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/866d4ab1df1c4824b426035957493a1f : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba718e535af146a59affb68c03ff33e7 : SUCCESS in 1h 09m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435be2958c534a39a5c4de82adfc27f6 : SUCCESS in 1h 38m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4c023e2396fc4cf4840c9e5a2a8419ef700f58d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-06 15:48:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f428024da0a3192ebc3c1cbd1720ab4a571eaa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 17:23:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cc5471aadfd4a8cb62875d83bdf0665\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497a261d68eb4cab94592f66b6cae6b9 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38bb26ba48d34e5498ea4152360942b0 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01e94ce614da476b9ef6bbafad74853c : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a29b307d1324bf0a607ab9a52cc91e9 : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/771d396571034bd8a742346e58550439 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/555497faa5f24ab69a1c2ae732eb2c28 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc98134691c2490aa79fe3999c5016bf : SUCCESS in 3m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/56a8d31c212a48f68957774866bcc241 : SUCCESS in 5m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6aea28b6682c4738baa8c8711b26c332 : SUCCESS in 5m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dad9e8bb9b424cccb17336f6441e6bfb : SUCCESS in 7m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3870b8f66d104bd693e8a9af644e2d94 : SUCCESS in 6m 16s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e1371b98d2244ce8a5f0c39e7347d017 : SUCCESS in 4m 39s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/724c09ae9b9146d48005eedec7eacfca : SUCCESS in 4m 29s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/d4479acfab254eee83877b625f40a24b : SUCCESS in 5m 07s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/29ebabe06b9b4f369fb9ca54b385ee14 : SUCCESS in 7m 15s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/43addb3944e94be58dff3b638534ce45 : SUCCESS in 4m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9c4925ef297c48ee91ea84ad0366b33c : SUCCESS in 4m 58s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/06a065b2fb904d6dbb698bec02b98ee2 : SUCCESS in 5m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/22fc9ef039b8440592542b9b2fb1e29f : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/219bafbbbfd54dbfa06e063e7091f06b : SUCCESS in 7m 00s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/608128662ed9402c9700f3e401fd4aa6 : SUCCESS in 7m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/913d6848f55646e9860ad8f3be20ade4 : SUCCESS in 11m 44s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7b57c47c54d94d92b24af17944f616d4 : SUCCESS in 2m 50s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/feb48910a7024415849a61c1f83f1aed : SUCCESS in 7m 48s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4a0555c1f1d94a6f889eb852427f699b : SUCCESS in 4m 25s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/6ab0fa1dff644f4bb44a9da6e133afc8 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5cf45e27229c4b04983acb8daf8cfbda : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4b5593c4ae224fab8a205167b2a49e91 : SUCCESS in 22m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/202d6e70c71d40e3bf04388697894324 : SUCCESS in 6m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b97c7e0d173044be9a063ebcb3c85e8f : SUCCESS in 11m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/661623a0eb904548b7ea3461ef6ea9d4 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3219adca650473a9d414804752fb606 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0558f7c92ce47349d19706211e4d864 : SUCCESS in 26m 36s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/66f41539c9f3461d8333d1caa7b7eaa4 : SUCCESS in 5m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/c4d47876c0224c72a942e725b9869b3a : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fafd3959f9554bd4b58d17ae8e9b567c : SUCCESS in 7m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/80fd7887aa1f47118ea7a197db67cfe1 : SUCCESS in 20m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6fb061f3a4e54f8da86ed5f1c78b7850 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06ada80f58c4479e849b25c79c47069d : SUCCESS in 1h 19m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/478aca2f561643b68aaacddf80e90963 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":3},{"id":"6d6edcfbecd4d7a60b02da653106342be64caca4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-07 12:01:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb416675e220ee87806e72d4bb4a7a1e5a40b0a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 12:02:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"72a043da0ff97f7820ca239ab38092dff8c3f1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 14:09:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f77c900dcdf4efc8451f81e65e1fd04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64df509fcbd847488bb16b65edcd6280 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa3fdebe66549b4ae37d8bd0320898a : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07782e6ccad4494798184aaf43567ebf : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf9ec74045fe4972824c604cab939324 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/078bce228db940baa34cc29c947f848e : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f581b7a1114a4a58ab8195ede9ea7a03 : SUCCESS in 4m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/72eb71f0fbbb41eba76b076da3d12107 : SUCCESS in 8m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1bf46b81c58d41628da82807be6987e1 : SUCCESS in 6m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cd4c8f79665543fc8b36d1865aedbc56 : SUCCESS in 3m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5c6c58d1743a4f41a84095f062cf3685 : SUCCESS in 8m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d83cb03ada234ca0b52f70305a05cb71 : SUCCESS in 7m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1153630071714df690438541a853b493 : SUCCESS in 4m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/db4ad3d72af6406b8bfd6d84ca650ed8 : SUCCESS in 5m 00s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9089d0829a08445b9b08bcab476e0f45 : SUCCESS in 5m 54s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3ca24e38b3b54bb7b15533122726c175 : POST_FAILURE in 32m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4310a792d4343049b634ac957dbd298 : SUCCESS in 9m 23s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/89da92da86db441f801e3f43fe03d9e3 : SUCCESS in 4m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dc487ca089e94b2ba06dc9107fe266a1 : SUCCESS in 6m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e685b135e62a420f91c12d8281663948 : SUCCESS in 2m 02s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/001c9f241ae4421c9c5f57fb86a52a7e : SUCCESS in 8m 48s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3fe23c38c38d4940b58a094ded8127ec : SUCCESS in 24m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/56221e64fc424f07ad7743db5df2ffc7 : SUCCESS in 6m 34s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/28ccb7d79bb948848a6ac75f9c6aa76e : SUCCESS in 13m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4fdba2b6be8d455b92adac12812cc070 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf852dcc8a3040c9b3ef7336c0b14073 : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3847c6903b174efdb0544385e10579aa : SUCCESS in 26m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4ccd0ec8be5c45d08129ded2649068d6 : SUCCESS in 5m 26s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e536423fe5174001bbbbf560ffab79d6 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b2cc846a4534eadbea8e77df13a4dc6 : SUCCESS in 7m 04s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/df454f2ec13c43928ddd2eed069f96d9 : SUCCESS in 18m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/37a9b9bfa0704406862866d1d1f616c9 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c12f542bbfdc443b82de346452a2af3e : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fb5dc9faec34f2f9c5346d51a4277f8 : SUCCESS in 1h 54m 53s","accounts_in_message":[],"_revision_number":3},{"id":"95cc0bcc60cc2e20469c36856be50fbed298c29a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-07 17:36:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"556b0663679cc4b82103d8ea39ea6a1b385179ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:47:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05db3154b1e34aa29442223b2fe55206\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d54364bba347baafe649188d0bb1a1 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de40521d2178475ba2e9ec77cca12e22 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5826ceab9f0043199dc3173961c3e83f : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/688a6f6b247845c59b879274daf5e41d : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/180ffac3706f494790fd641c589a3a94 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7ae4f9db51a4c26a15369cdb305b725 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d36e997414814ba9968a9da97a6541ad : SUCCESS in 3m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ca43029ebac54f0697366a8584c95d8f : SUCCESS in 5m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c592559b10c443db92e2c8727846f49 : SUCCESS in 4m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec66aa9d2ed145f98c443e603f77040e : SUCCESS in 7m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/934c4cfda84041ed90cce0f64d62f400 : SUCCESS in 5m 24s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/728ab20dc4294a448c270d26722a9f83 : SUCCESS in 6m 11s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/dd24438037ca4230be7d44a4391f5cb4 : SUCCESS in 5m 06s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/883ccab800eb4792be233107885ff865 : SUCCESS in 5m 24s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4b96e3a95781473c91cc23f9506e2083 : TIMED_OUT in 40m 48s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a6a7f0a59eff4bcbadd9a3c29357961e : SUCCESS in 6m 08s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/bee0ea04e0924061a4c34c7f0040c587 : SUCCESS in 4m 41s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b048c16fd2ac4deab5df16da53b6d7ed : SUCCESS in 6m 08s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/611f1e620c7e4eaf846a9007b919e2a9 : SUCCESS in 15m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b089344ac59c4c68b06584dcc8268ce2 : SUCCESS in 7m 40s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b0273b5eb3e4484fb3cf8a453fe95369 : SUCCESS in 7m 08s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c8d22d2222d74061ba015a775d5d134f : SUCCESS in 8m 05s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e49e802c18b149d2bbca993c6d5a58a1 : SUCCESS in 2m 14s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2c032190e68c4a2f834680a9e1386662 : SUCCESS in 7m 47s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/abf070dd4d924de2b94e1776aba4bd7b : SUCCESS in 3m 51s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7d5f81e0443c43efbf441989652b5ceb : SUCCESS in 3m 46s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/45ffcee65d59496ea139ea1e719be01c : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dad179f5715649ba8949ce9289a5a765 : SUCCESS in 27m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7e918580c250447cb38b1ff25271b963 : SUCCESS in 6m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/03864cc5ec834eef9061244d0ac62a75 : SUCCESS in 12m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6b4d5a28ea414da98773889ad327805e : SUCCESS in 3m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ff5dca4a3264718a629f98ebb1e2ffc : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9507960637d340548bbeaac415472411 : SUCCESS in 29m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/aaa79872dc9e46788497d610680105e0 : SUCCESS in 4m 57s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8c1cfaa3f72c4e148aa79b066655aa51 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1ab8fa4bf0743b1b37943b56684e355 : SUCCESS in 7m 34s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/942783739ee449469f764e04ec1cc152 : SUCCESS in 18m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f3057cb17fdd4b9fa2803b7611c7b182 : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7f38b3794b84ff2a9b9fc552226b17a : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/872bf831f0a7450bb54b7565cad650cb : SUCCESS in 2h 04m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d92009d76cae5ea7f8da119b8ff332dcdc491584","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-07 19:50:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ff551e48bb85f8f9537d5ec0721398efe7bd3ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 21:51:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eba6ddcf639401dbc25b16bf9267f70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf300122c0c453fa6fadb02a8ee30f8 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3cb64afa954a56b7e6cf46e5f6d57f : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f5d4380bf3848399b88fe17188302ee : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3dc516875ebf4e27afd11738cd4dc92e : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/920c7805f37e401eb593aabdcd6dc105 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebe1b4aada8c4a1db73d0e662ad6177b : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/495ed2efc558481086eb1e77f8afaf61 : SUCCESS in 3m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3990c8c7b35e4902b7d01cbcfae04ae0 : SUCCESS in 4m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67271274406c406ba951176f11a0962a : SUCCESS in 5m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0c3f6b1c4f0e46b480625986af208715 : SUCCESS in 6m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3105c7905b68447b8e6026e0116802ab : SUCCESS in 5m 30s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ce96de7ed8ef4413a6ef16a8dbd7b70c : SUCCESS in 4m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2bf924ae4d574e2c8898dc166c13565a : SUCCESS in 4m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/fb341a01299b4d34804496c19d6c3afe : SUCCESS in 5m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/125852b3ae8644a088edfc97aafbcd32 : SUCCESS in 10m 08s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c2924e11973449ed87335dfc798c02f1 : SUCCESS in 4m 13s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7606368b59d544b68d88759370b1c701 : SUCCESS in 4m 15s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/07eb90ba21fd406290e48ac3acbaba8d : SUCCESS in 4m 43s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1bc3009e1ad448cbab6dc2394d609f0b : SUCCESS in 15m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6025f9407b04501b0a501a0a429d401 : SUCCESS in 7m 53s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/78787c5db9d646249da6abd83bcc4018 : SUCCESS in 7m 26s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dea55a89b3fb410aa4924d8f09458131 : SUCCESS in 8m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c32e96cec20c4c038e474ac887468ad4 : SUCCESS in 2m 28s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2043d365a9f645dc9fa245ca077e6bd4 : SUCCESS in 7m 56s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/40d2124ab8674e508a72d98bf183f63b : SUCCESS in 3m 18s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c0e7d86322af44bdac665b5bd03382c5 : SUCCESS in 3m 54s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ada5354899f540f0a4ef667c35421367 : SUCCESS in 10m 03s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/bced1e04b53c46caa1db5a75c1b10609 : SUCCESS in 25m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f2cbe656348a4fc2a5f451ac956b1ac5 : SUCCESS in 7m 28s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ba5a1ea37a964054ad39b3f9132ada6e : SUCCESS in 16m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7119a3f9085d42b3a27741d462f495fe : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e20d296b2b04d5084c602c850f6f282 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd2c255ee0084b188f9488300b8eb7b7 : SUCCESS in 26m 48s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c5b8319067bc417ca42971d2a7f0ec5f : SUCCESS in 2m 50s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7fef86b9d58244fbb4d831788306d343 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/107fd38c813b4184a8a6fc90f7d68cce : SUCCESS in 4m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4ca694dfd45b48f2b83947aeb857007a : SUCCESS in 16m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f689238d0db84494a1759785cc7ebbf7 : SUCCESS in 6m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a67cac2fe96482694b0ab5379903cc9 : SUCCESS in 1h 05m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85fcf1c3be604c1f8b8e91bfa8815431 : SUCCESS in 1h 55m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1788b20fe7dfa7464d2254d09ed343492c02b07a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 21:52:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c5bc33edc08ac29689aaf2a5c532d53ccc9f8fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:16:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50be133ef77643c4a0615dcda4eb25fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fe6fc7c85f47d080e1556ba38e9ad7 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5780ce3be640491abf3c879382a49a31 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ad50f50f3bb4a50bf552634ab68f09a : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85371699bfb456aa7a0ab2d5f1925ef : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac40bbcd1a8649e2a3901d3f2e4e6ad0 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43d0a4cbf84a441d8f8ac170802a706b : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa9c3ff171fd455dae2a58e1fad0d7c1 : SUCCESS in 4m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bdc8d3082cfa41789e6df1437986f1fd : SUCCESS in 26m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f4252bf325344132b08072cd641c0799 : SUCCESS in 30m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b9076975199a4bf6a22a465d249c3cc0 : SUCCESS in 4m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e1f10fd7adeb4e2f99ab35e5b520f288 : SUCCESS in 10m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b293a85029364eb29729bbe41209f18c : SUCCESS in 3m 55s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/43b3f52c6b5b4bc8b82f5ec5e16ad052 : SUCCESS in 4m 07s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/23ad702093d244a4a3684eeb3a83836e : SUCCESS in 6m 29s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5e14f3b9e8cf4b7eb40d7c12fc7fca58 : SUCCESS in 7m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42c8a553212e412fb9fb8e850039097d : SUCCESS in 6m 08s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9caa715b55d04e9ba4e806042047c8b2 : SUCCESS in 5m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/78ded0e4b371481f8cba55b775c06a81 : SUCCESS in 6m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d567a9cce79d4fcb90c267d4edcdf617 : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/be10af99374f4fb7888b87dcfa1a7a57 : SUCCESS in 4m 21s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0d7c1c54f9da49169df08e8cffc91bd8 : SUCCESS in 23m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f4c2c99c361412c9336a7705b6ddd63 : SUCCESS in 6m 34s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/145ce128f6b242279e42a3a2a77e9061 : SUCCESS in 14m 29s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/13692fdde84b48bf9e15e0ffdbcca37a : SUCCESS in 2m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25645b5aee694b0da470ff491fe5babf : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47f0f15027f4401490668c6ed725fcee : SUCCESS in 24m 09s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0e41c452fcb142fda498eab09acdbd4e : SUCCESS in 5m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6aa9ad82391f4020a76e59ca05acf077 : SUCCESS in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80a3ddf814504298880126f89009f0cd : SUCCESS in 6m 46s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/02a4d6ed75454eb9b4508882f41df378 : SUCCESS in 12m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5396dc45284247e89f6d1809aff99be9 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e2d503f673049faa1e6ca5e0c5b08fe : SUCCESS in 1h 17m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8f33db7d95a4cbc9a363b509603c754 : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5a7a7f875f3ecf6e8337fc4ac46c0a440744aa94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:16:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5990f0d8836ed8a2fc1aa500dd4df1283af40145","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:17:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/828be2554c6846e896aadcd62465bef3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f02ace74dc64448b6cd1028cf5a62ba : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"193272fbc4572f5a2d06d6bb49277adb9cfa8721","revisions":{"05ae2927a64098a8e423b47a2625433df35db591":{"kind":"REWORK","_number":1,"created":"2026-05-02 08:18:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/986968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/986968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/986968/1"}}},"commit":{"parents":[{"commit":"ed679515f6e053d20b05ffcc30b3172de864f7f1","subject":"update constraint for oslo.privsep to new release 3.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-02 08:18:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-02 08:18:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I26073c75f5fba0c869854ce02406e2673f3ca881\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ae2927a64098a8e423b47a2625433df35db591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ae2927a64098a8e423b47a2625433df35db591"}]},"branch":"refs/heads/master"},"87265a716bb911f1c474fbd1ed40994dd08941e2":{"kind":"REWORK","_number":2,"created":"2026-05-05 12:22:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/68/986968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/986968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/986968/2"}}},"commit":{"parents":[{"commit":"ed679515f6e053d20b05ffcc30b3172de864f7f1","subject":"update constraint for oslo.privsep to new release 3.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-02 08:18:12.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-05 12:21:50.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I26073c75f5fba0c869854ce02406e2673f3ca881\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87265a716bb911f1c474fbd1ed40994dd08941e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87265a716bb911f1c474fbd1ed40994dd08941e2"}]},"branch":"refs/heads/master"},"193272fbc4572f5a2d06d6bb49277adb9cfa8721":{"kind":"REWORK","_number":3,"created":"2026-05-05 16:46:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/68/986968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/986968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/986968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/986968/3"}}},"commit":{"parents":[{"commit":"952fd8534c4e03da51f7ca09a9519bc7ff935508","subject":"Merge \"update constraint for sushy to new release 5.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/952fd8534c4e03da51f7ca09a9519bc7ff935508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-02 08:18:12.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-05 16:45:48.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I26073c75f5fba0c869854ce02406e2673f3ca881\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/193272fbc4572f5a2d06d6bb49277adb9cfa8721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/193272fbc4572f5a2d06d6bb49277adb9cfa8721"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
