)]}'
{"id":"openstack%2Frequirements~987207","triplet_id":"openstack%2Frequirements~master~I18ece4022ea51ec06abafdf91086d1ce945ced43","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I18ece4022ea51ec06abafdf91086d1ce945ced43","subject":"Bump awscurl","status":"ABANDONED","created":"2026-05-04 16:42:53.000000000","updated":"2026-05-20 07:54:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25d404a370735596c9975993a34bf5c470775190","_number":987207,"virtual_id_number":987207,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-04 18:41:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-04 18:41:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec1b03597630c4f66d4f6aca158457fbf9b5e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 16:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfdfd17dee30583ceab4a391b0c90ebb1cf58e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:41:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daeb4393b9134afca1e7b6a59d534e38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a237a3fb2fb04ddfa39f960a358e600c : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd63b9498d594346b93642e3c6b70091 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bb9d288310d40b69462a5c34d227144 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1aad419dbfac4c5585978f0cc2f72d5e : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edf1ce57102b4b4da44a53a2209cd7e0 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/551f9e1b0167429db5dc3ff198332362 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbd247b5ba6b43cab4268fa70fba574d : SUCCESS in 4m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/973eebf7d48748fda9cd2b416b0e7064 : SUCCESS in 5m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1dc229c642144bd9911834cf4d17698e : SUCCESS in 10m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d6b841c09dd4b58b9227b63a77ffb65 : SUCCESS in 5m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/70cb2b5bf3c64ec49d78524b2f5a2bb7 : SUCCESS in 5m 57s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/08b503680d254024bedff707a9998c6c : SUCCESS in 6m 33s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3a80441d8abe46bbb91818e69a28e4db : SUCCESS in 4m 50s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8040f7e934324ad29cc82f57ee4858cc : SUCCESS in 5m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/85192e5fd645466a9af29308afac4a35 : SUCCESS in 11m 17s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/dea39eba29274a4fb11bb48bd1b727e0 : SUCCESS in 4m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9c07a7b94b864278bd274795983864ee : SUCCESS in 5m 04s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2c53ece8b75c4924a3c96422e996e186 : SUCCESS in 6m 48s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5951e9fad3b0404fa87dfdeccc8e21bf : SUCCESS in 15m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ebfcf2d89aa4b30b2808bc4a0951ffb : SUCCESS in 6m 33s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0a271cea9d254d70b39f9e32160a4fae : SUCCESS in 7m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0c2a7c2bdc4b4e8e80f5bafed8471261 : SUCCESS in 12m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/48d0f131160b434982c3259376a163c7 : SUCCESS in 2m 29s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/acad4de873d54fbc849ee9e25affeff6 : SUCCESS in 8m 18s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/fca110ce11974d16b25aa82bc33d30eb : SUCCESS in 3m 56s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/fa355285ef6f414cbdfa4616250997ed : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d21e2ecefebc4c9aa20ad655469cdcac : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/362b4604e5d44e55ba8c99ac02bbd3d6 : SUCCESS in 22m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fc445c25f85a45bbb4a8c75f051bfa40 : SUCCESS in 6m 36s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/68948d6f651148fea7e88f68020777a8 : SUCCESS in 14m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e650c69fe2354abfa2476be37517478c : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9de060137aaf4c078a53d3aee369a7b3 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b93911c6aaa14f5d98c9eaf4290cbf4c : SUCCESS in 25m 39s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/10dec1f25255454db92ac70dda95249f : SUCCESS in 5m 25s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2c74369d56794ef584c6c01a79755733 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58ed4e2eb2064e8d864c35ea79e71736 : SUCCESS in 7m 03s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/cb6c67a83f1a4a9aa79e2ca3ae72b748 : SUCCESS in 18m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d3eeb7f0b0ea455196db7fd9012f197a : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/188b40af001946f08d8b7b0f0da71440 : SUCCESS in 1h 04m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c5ef39775c4f769370c71588669b36 : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"25d404a370735596c9975993a34bf5c470775190","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:54:01.000000000","message":"Abandoned\n\nthis bump has already been merged, thx for verifying the fix","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3a7eefb03f522c9ad294aa0d588fab37a7a548a","revisions":{"a3a7eefb03f522c9ad294aa0d588fab37a7a548a":{"kind":"REWORK","_number":1,"created":"2026-05-04 16:42:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/987207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/987207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/987207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/987207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/987207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/987207/1"}}},"commit":{"parents":[{"commit":"ed679515f6e053d20b05ffcc30b3172de864f7f1","subject":"update constraint for oslo.privsep to new release 3.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 16:42:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 16:42:26.000000000","tz":540},"subject":"Bump awscurl","message":"Bump awscurl\n\n... to the latest version currently available.\n\nDepends-on: https://review.opendev.org/c/openstack/ceilometer/+/987205\nChange-Id: I18ece4022ea51ec06abafdf91086d1ce945ced43\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3a7eefb03f522c9ad294aa0d588fab37a7a548a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3a7eefb03f522c9ad294aa0d588fab37a7a548a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
