)]}'
{"id":"openstack%2Frequirements~987339","triplet_id":"openstack%2Frequirements~master~I32c040a35f2603ae6579aabaa146f4282e9f6642","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I32c040a35f2603ae6579aabaa146f4282e9f6642","subject":"update constraint for osc-lib to new release 4.6.0","status":"MERGED","created":"2026-05-05 11:57:05.000000000","updated":"2026-05-05 17:59:47.000000000","submitted":"2026-05-05 17:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987339-new-release","meta_rev_id":"da910f0b9fdf5ae5c1d405984a49a630a54b23c7","_number":987339,"virtual_id_number":987339,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 17:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-05 12:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-05 17:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-05 12:42:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 12:42:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-05 13:58:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 15:15:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-05-05 17:35:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1ec4aa2f61d5c20c17d76dd133a3b17ff526c9c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-05 11:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59531fbc104b88f0420f1f425236c0bc180d940e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 12:42:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f641581b0f6210a1b71fe8802fbcdc0385383236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:58:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ef92696a7347f3a24f159d21522ad7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0811b128b2c64255a96a6340edc77d54 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc8ddf30de945a5b7f31454c18eebfd : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/885d0b1e7d6b402fa3bf4db12fa317db : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa719fde6d8041979016d378cf704ffc : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55e88a22074542dcb69739d983be61de : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/812e12079abe496c8705fcdbdfc833fb : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02efae8023464762a63538426f2ae775 : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3b9f07f7fba9461faded8d18351052ad : SUCCESS in 4m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b3bdf4e2bd234be6b7a0c75f5581a85c : SUCCESS in 3m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a8b4756e3d7a46bbb282b5d8e8a33010 : SUCCESS in 5m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/af5c875ef17f4c04a894b062d8f5b271 : SUCCESS in 5m 37s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d8c4ea2caec94c61808831d93644a09c : SUCCESS in 3m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5d83784f461d41528026ea69bd21ac0d : SUCCESS in 4m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/52d6d7107ae543aab429a03dd1bfb01d : SUCCESS in 3m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/1f856ba08fbc4afea7ff83edf6262a31 : SUCCESS in 10m 21s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a6e5aafb2fa84f0897426a9aa15186ed : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4a3846c905f1486a9a45fa1c37318ed2 : SUCCESS in 4m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0191a61283004d1f996d7c5b2cd329bb : SUCCESS in 6m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a7776710cc964eccaeea5d07c7395622 : SUCCESS in 15m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d934d0466134b69b2a37fa32ef2f49a : SUCCESS in 5m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bb3eb68c4c244550822480250890dfcd : SUCCESS in 8m 19s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/fba7d305a5704aa4b938b8cd911bc35d : SUCCESS in 10m 06s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/46747caa7abf41cbb4c574e450b3cf5f : SUCCESS in 2m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d24fe984556b41a6b1818d63496f396b : SUCCESS in 7m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/cbac145ce20645a29d2f20a8ed70ba7c : SUCCESS in 4m 12s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/494066ee466d43da8ca3a612ada628ae : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/afd3897cd640449584778db948a11ec1 : SUCCESS in 9m 31s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a477f2426ae445ea9aff72f80f69b48b : SUCCESS in 19m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3e7e33a278d54a72a6bc463da59e351b : SUCCESS in 6m 01s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/369a66c1329a4134abe0652e250fda0f : SUCCESS in 10m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/40e7b74b0a1f42419e341e406d5c6943 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac8c3677af8b4e608d6f384aff7d7927 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d51c89b4d09649bc80b2690a5380f66f : SUCCESS in 24m 42s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/cc0a6549558f414ca4ec0b73c8da7c6f : SUCCESS in 5m 19s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5f489791db244d3dae539a772425567d : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf0b6bc5395a4f3c8ac037f8b810da15 : SUCCESS in 6m 39s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/516781f7466948ce91eb2d5b0647dbf2 : SUCCESS in 8m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e780d77216c74e1f8f353bd94f0b0053 : SUCCESS in 3m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8a24c56e3c249058524eb5ac17edf4a : SUCCESS in 37m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad578e005e03423b8cd034f24fab71dc : POST_FAILURE in 1h 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"75e62c4e22ca236e08f78c3cc42e5da6cb5d08b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 15:15:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3aeef44848200dd7415882508a526d92eeb0f225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:11:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43dae90386f45d4a73b432c403181ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b905c5ae469641eabc2c962a1ec4dee7 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79396c11163c4ef0abff48af51736600 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c8604424b624b2a917748108d7992ff : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cf4cb977cdb449abef416685c67c49c : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9faffed094454822b415c8da6683415c : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/654e63c8878a41fb95cdfd3bd9f8d7df : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/075a5e56f350492892cc27f685fcc1b2 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef13b3a49b5b4ed094ad001550775958 : SUCCESS in 4m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/36a962b95b6e4501ba05f014fa9fae64 : SUCCESS in 10m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2b6cc573b99845b191dcd93b5d3a9baf : SUCCESS in 6m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a211fb63174e410f9936edfd0696a791 : SUCCESS in 5m 17s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ff791068be9f4b608ab9c0208bd84643 : SUCCESS in 4m 15s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/742d65cfdc214c219b0e53562cca5c67 : SUCCESS in 4m 50s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/7bc14e2022614a90966cd18244f571a9 : SUCCESS in 5m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f12591ece3ce45b5adcfe6df417f5096 : SUCCESS in 10m 52s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c755e97e739346059610beeb9d78a4cf : SUCCESS in 4m 42s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ce3d7ddd827244fbbd3ebe4d1e54b386 : SUCCESS in 2m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/916754fec1dc4ecbbdbe355dc75860d5 : SUCCESS in 6m 19s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/77587ec616a54dbdb3fb19ed890a960a : SUCCESS in 15m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9bf82863ef040b3a7f1a39dcca88ad3 : SUCCESS in 6m 44s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4775ad5b0b5b45eabcd208662951a88c : SUCCESS in 7m 53s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a6dd474bf5d34b75847c65a85aef8e9f : SUCCESS in 11m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b5cc89672f9d42f2bf1a2c7b252a1a70 : SUCCESS in 3m 51s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8ccbf9d042674adebef177ec728d2ce1 : SUCCESS in 10m 00s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9e39cc5bfce94c8fb4dada57d4e62f97 : SUCCESS in 3m 35s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/6ee8bc93ecc24328be6e2bc4fc9bb10b : SUCCESS in 4m 02s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f087c2817f5240209b90d582a27fd2fa : SUCCESS in 13m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dae075b603774e5daeb5dbf548a7aad8 : SUCCESS in 24m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92c800797bb548e29a72641c67c6189e : SUCCESS in 6m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e1eb584d0c9249a7a64ba0d8c41f7d32 : SUCCESS in 13m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8fa5ac19e9e1417ca9b10b660dcdee5a : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0276da36e5846108f76972211bfc61e : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b547e8dc0a1445e3999c7412e8155092 : SUCCESS in 24m 22s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7284b58492ae42cdaaa9d33edace6720 : SUCCESS in 5m 37s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/54cf610051364b93b210f7a7335c3863 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/766058a2e510459ea302a3b194142877 : SUCCESS in 7m 47s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1c92d1e2bfdc40f7a8024989825c707a : SUCCESS in 20m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ad8f79a9acad49ccbb35368a7e3ecfd2 : SUCCESS in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5941af3000a949428c60936a8e8146db : SUCCESS in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b389571295ba43ec9e2eefcf4de48b87 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ff870e0887aa7ec28116236d6689481c83850d64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:12:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1dc98825d8cb402bac16ad310c5598c9354358a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 17:35:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"039554c89896c37fca68d634bf6738eb1e0d132a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:58:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7798881732db4f438f1f8207e3837c4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711ba307a90448cfbc7f0c3128c01703 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a6d2be41cf43ae9be568844c8bfb4d : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/518232a29478420282f4ecd5743ed590 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfa0619f19cd4b51a901135443ce1254 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e50ac632a9324ca2a23ba2fca49a2f30 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee3bab5ee232458cad297a2145b22b35 : SUCCESS in 3m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/501fece75fd74a40829f4838a10d8878 : SUCCESS in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/999aa680ec804c21b601b416c546e40e : SUCCESS in 7m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/35442aae2506461d88d8e8cb1f81fce3 : SUCCESS in 6m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0dadc8f60eee4dc7aea638792d97ff3c : SUCCESS in 9m 38s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/86e4ee5a67eb470b8e9c7f531510dd68 : SUCCESS in 10m 18s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/274d5af804fe45948b7950e2671c6dba : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/57583785b13941dca8299216339b052e : SUCCESS in 5m 14s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/62ac9481f12343afb10e6c06c9787e87 : SUCCESS in 4m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/99c0e307edef4035b749706026cd9201 : SUCCESS in 15m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6bc0e0214b04f368831bb6adb2ab947 : SUCCESS in 7m 50s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5bc715fff27742deaad3743651409bde : SUCCESS in 8m 41s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0e449f9a781549ac83dae88259ddb911 : SUCCESS in 11m 53s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c4af76cca42d4f8e90a13ac4c5090598 : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f137d53be85f4c448becb45b3a73c9a1 : SUCCESS in 7m 23s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b720141861454c769c43485eca0e138c : SUCCESS in 21m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/78450cbd79124ac1a39e59f769f96721 : SUCCESS in 6m 38s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/aaeab4c2db864d72a8ceef1b72f5755e : SUCCESS in 8m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3660246915ce4eb394a6b0f808b0cab1 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74d68b42b3014aa692326da1dcff0452 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd1adbc893b84a418bf9377a6ae196fb : SUCCESS in 25m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e7c841d1306e47fb9f916d5a83cf3d67 : SUCCESS in 7m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/907ad712ab9f4b1996b40580e0dc347d : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/617a03c3e020456ab8abae1c844e08c8 : SUCCESS in 7m 29s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0eeff9adb9ef4b5aa4508053099aeb46 : SUCCESS in 15m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/55d1b160d9ae42d9abe5d9601b8b92b6 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae6931c0a63b49949f1a5d4a960d8fbe : SUCCESS in 35m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5f6eebf7ed74b9786b0f3e2ae907164 : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"746234c07bc3df10a82d2c971773cc694edf2d94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:58:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da910f0b9fdf5ae5c1d405984a49a630a54b23c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:59:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77d81a6360040b7a2d2db2862270603\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0407fc933b4aa7b59597c989a14875 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd286afab8c8fb05d5c0a6a3bb5ab13cd3335834","revisions":{"bd286afab8c8fb05d5c0a6a3bb5ab13cd3335834":{"kind":"REWORK","_number":1,"created":"2026-05-05 11:57:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/987339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/987339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/987339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/987339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/987339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/987339/1"}}},"commit":{"parents":[{"commit":"ed679515f6e053d20b05ffcc30b3172de864f7f1","subject":"update constraint for oslo.privsep to new release 3.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 11:47:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 11:47:04.000000000","tz":0},"subject":"update constraint for osc-lib to new release 4.6.0","message":"update constraint for osc-lib to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie841535405d1b2c0405df41c0d42b9be8e359421\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nChange-Id: I32c040a35f2603ae6579aabaa146f4282e9f6642\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd286afab8c8fb05d5c0a6a3bb5ab13cd3335834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd286afab8c8fb05d5c0a6a3bb5ab13cd3335834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
