)]}'
{"id":"openstack%2Frequirements~987345","triplet_id":"openstack%2Frequirements~stable%2F2025.1~If1a100251a55d088c1470699fee8680a39d5f175","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"If1a100251a55d088c1470699fee8680a39d5f175","subject":"update constraint for sushy to new release 5.5.1","status":"MERGED","created":"2026-05-05 12:17:30.000000000","updated":"2026-05-05 15:56:06.000000000","submitted":"2026-05-05 15:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987345-new-release","meta_rev_id":"92f89886c262a95e4b316622e50d8973fdc95fa7","_number":987345,"virtual_id_number":987345,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 15:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-05 12:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-05 12:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 12:43:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-05 13:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a51c217ba9ef38934ace49c70f75cf7d259ecce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-05 12:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b23a53f21f87b8b28d5d6f4295620c6f32aa61f3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 12:43:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1939fb55eb5be904d73d381b86b31d63503d5bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b4e8a233ab4a9cbf05efd6c1db3401\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1439d4f04144640a0ba6aed201b98ac : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20ea6281f8db4223aa85719e53b41564 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4aa37bdfc1ee44c8b4b105ba061dcfa6 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23562a4c61a242e1be7619f28415a9a2 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba5c6d1010304233a9a80c4138bc563a : SUCCESS in 5m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3cad716d491f4fc0bcf21815007a0485 : SUCCESS in 17m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5583a1174c8741bb937ea37f77ff045c : SUCCESS in 10m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c8506733396f48f0978962f55f3f468d : SUCCESS in 18m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/63cc16beff5d481caa38b4d1311b683b : SUCCESS in 23m 19s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/47051b65aa7547259d4f0efa8cb573ad : SUCCESS in 4m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/05ac7821e106470eae9fdf206f33c3cb : SUCCESS in 6m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0a4cd204d1b34f0d92ab51055749fd97 : SUCCESS in 2m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9e1d1a079dce44a48122e5cdb9cdaf71 : SUCCESS in 10m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c6ea8dc01e594a979d51b6f4b9065cfd : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/11192b462b6d46a0845b702cf0f969dd : SUCCESS in 3m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f124e3ea718f485a85e95f4ec233dc76 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1b6fd6eeb20f4436b57d4c5b9dfd0cf8 : SUCCESS in 14m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37e20d0744b942c6bf9d9b2918b5e0bf : SUCCESS in 7m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a534f38413bd42118bdd5618fbc2eafd : SUCCESS in 6m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c261b20c955e45f49ea06269cd4d2ea5 : SUCCESS in 13m 32s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c5745a9484e44cdfba33be6fcb0c52d5 : SUCCESS in 3m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d3df11b6caa94e65a8d8c5a190dd209c : SUCCESS in 9m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b4988ef8147f417690cf2b188cc4bccb : SUCCESS in 2m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f144885ae27341ce86c5a85da67e02af : FAILURE in 4m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/35ab177f16ea43d18dca8a646f4f53cf : FAILURE in 4m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/be1db43c1a3f4beeab19a7f17cc58473 : SUCCESS in 21m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6463d649c2d3415b88c1d53e0661e24e : SUCCESS in 4m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0c913d85cdb9492eb147f05460ce814b : SUCCESS in 14m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3a5473f2f4814485b40a34ed34f62d53 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe16f60030404690bb09ba6e3534c83b : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2ea6d176ab04609b52cc07f61708e9b : SUCCESS in 27m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8b8aefe830f143ba92043861db05a743 : SUCCESS in 7m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b20fd00f6415429c98d93bb09c01d67e : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff6900290cfa4490909e4b0d534fc09c : SUCCESS in 6m 50s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1d34a343b88e4215a706a4bc4f35e417 : SUCCESS in 18m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8704157ecaf84fdd8a1fde3afdae861a : SUCCESS in 1h 04m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4727f8061d0d4e7c824a9e84b0d373c5 : SUCCESS in 1h 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"987afe2c73d9eaec8246d9b0df46e9623ea01f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 14:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b26c0ed798a3ef11b1830b1389fd8a4c7d98e0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 15:56:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7847ffabba4ee4a133e583d13bd956\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db7508ceae04e7db2cdb10e3cc59cf2 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0e198ead96e492d97700db5f068fccc : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94bc34046b76445eb1322aea6fafcb80 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7c9a674000c4ff4a7d237d077f2b8ee : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2cc93fdcb5b45299d41c7313d45f3c1 : SUCCESS in 3m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1c79d1b14aa1411293db7241f28c35cc : SUCCESS in 19m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4bad62355c8c4a9f907f08aa5eec78fc : SUCCESS in 9m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/71d66b3943e24376b214e06855d67b2b : SUCCESS in 19m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2a98b7f7387148bab01cde0fa97bed49 : SUCCESS in 12m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1f004e9f2fda41da8dc3a7dfa679d92a : SUCCESS in 12m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/40d427b86fa2446390097f1f7cb1739b : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dc14754b22084517a14609ff991a1b7a : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c6e834ab45884a27866037a43602e2c7 : SUCCESS in 8m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c7b9d85a76c34864a7d943ab4e67fb40 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fccb993dbfa4074b4ba07a840029dfd : SUCCESS in 7m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4223b42aeeca463bb9d75eb1a67ad16d : SUCCESS in 10m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b1874a817c21467eae6e855f8a017330 : SUCCESS in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6e8f5589a0c44c748ee37bda5bb88c31 : SUCCESS in 14m 02s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/29a80f0b163e44efb8382e398413a690 : SUCCESS in 22m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ead5994809974ae288d7d445099bfe0c : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6c6ef052a6974f8e83afafc6b58fff86 : SUCCESS in 14m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/64aea9ea177740d385811a8ed57614c2 : SUCCESS in 7m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52d0aa3413f04a61b4b4ac67da9584dd : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a72849fa234644afb25cb24c5a5e5c37 : SUCCESS in 27m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6a7c51533bd842d5abbeca60e375d91d : SUCCESS in 7m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0c8bb34d00f64ca98c11fab008093223 : SUCCESS in 5m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/219a21a757734c5494fb4ba1a1e88f04 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2bb97b9591947c99d24211fa6cb52f6 : SUCCESS in 1h 00m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29e8fc6b24fd43e0b2c14ece95217374 : SUCCESS in 1h 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"92f89886c262a95e4b316622e50d8973fdc95fa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 15:56:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c65c03328454f762ef0e173f77b1ed05e1d80937","revisions":{"c65c03328454f762ef0e173f77b1ed05e1d80937":{"kind":"REWORK","_number":1,"created":"2026-05-05 12:17:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/987345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/987345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/987345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/987345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/987345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/987345/1"}}},"commit":{"parents":[{"commit":"c10699eb2e1de27287feeb35ea29db2cf58bf4f8","subject":"stable-only: drop bionic wheel cache job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c10699eb2e1de27287feeb35ea29db2cf58bf4f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 12:07:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 12:07:29.000000000","tz":0},"subject":"update constraint for sushy to new release 5.5.1","message":"update constraint for sushy to new release 5.5.1\n\nmeta: version: 5.5.1\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ib4d3080813b3b006361d91662c5ffab1f2f3d1b1\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nChange-Id: If1a100251a55d088c1470699fee8680a39d5f175\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c65c03328454f762ef0e173f77b1ed05e1d80937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c65c03328454f762ef0e173f77b1ed05e1d80937"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
