)]}'
{"id":"openstack%2Frequirements~987353","triplet_id":"openstack%2Frequirements~stable%2F2026.1~I392f2a3e6d95bb670c3dc1bb5fc68feb2cbc87c1","project":"openstack/requirements","branch":"stable/2026.1","topic":"new-release","hashtags":[],"change_id":"I392f2a3e6d95bb670c3dc1bb5fc68feb2cbc87c1","subject":"update constraint for horizon to new release 25.7.3","status":"MERGED","created":"2026-05-05 13:05:38.000000000","updated":"2026-05-05 17:58:54.000000000","submitted":"2026-05-05 17:58:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987353-new-release","meta_rev_id":"89b0a602c8f8d96ac253187704e5c62ccdd169e1","_number":987353,"virtual_id_number":987353,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 17:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-05 16:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-05 17:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-05 16:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 15:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 16:43:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-05 17:34:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"681c7ad50441f973072f884450ceff97e36362ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-05 13:05:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93bb7b3a90b04cbf9e8a6ba5ac21c59c270a8e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 15:07:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/becaa41c5bd748e19ef52a59ae6ffce1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/970b6536a2f44bb3976a7d6322055ab6 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a11a7f0f581a49b39db5e454b591c859 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0749f7c36ed46d783f2595f2172ed7c : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/473550523be642c786aff198a0c6f4e0 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8374b9736fab45dcacf96a3a3ea7b765 : SUCCESS in 3m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/58a620ab3eb34078b3a417ca671f0408 : SUCCESS in 4m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3e2158655cb84cd9b291406d756c75bd : SUCCESS in 9m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d1d79d038cc4141a4ccf1123f834803 : SUCCESS in 5m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c3edf19d3aec4ba68a0e4aaee8fcb06c : SUCCESS in 7m 40s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7c66409171e248bba848b41c36f70703 : SUCCESS in 2m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/63bf04e40bed45cbb8e559f1de5c3a1b : SUCCESS in 5m 20s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d6cc72a59606425fb36b99e4d323f609 : SUCCESS in 5m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/62f3325151f24679bd18bd89de7758ee : SUCCESS in 11m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/56842a04f44b41c5a69fe5473c65f3f4 : SUCCESS in 6m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6cb21fd331324f14a850f0089fcd7a36 : SUCCESS in 7m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/51bf1913811c46a08e6798566b25ef8f : SUCCESS in 8m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bca92fc83de643b697bfb988105c3e9b : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/889cf6dea6664fd59563937b5da907e6 : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1237bd7fff8449b6a95b88c8e72166b4 : SUCCESS in 10m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/de9e3cd0379349148747299714bb05af : SUCCESS in 12m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9825a637a8694451b4e7f0e418e22389 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5d34ada735b8463ab507b4e3e7fcaf84 : SUCCESS in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/43df9cb28cc546509c10d4516d913ebc : SUCCESS in 4m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/77c5dd815bef4f9fb9b46aa2652e574a : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bf863e0191e74b0bb2d264676174b300 : SUCCESS in 9m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/44592000d7a0422d8487fad568a20d2e : SUCCESS in 22m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5baf6592ce704f30b762f85b4e8f697b : SUCCESS in 9m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b8fe76f10fa54017be8aa568815eb64a : SUCCESS in 17m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3cd585cb8ce44bceb538b63da8dd583f : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5955847474844b37a575642640dc3807 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d65366d7514c485da693e3ce40535d0d : SUCCESS in 33m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/513e189890614f109f236e31dd571c0a : SUCCESS in 5m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/530092d1233e44b8a90d2e459915a681 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15c43e4dd6f8439e8cc450082929ca15 : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/008cfd6974b846ada7f714a783a9a7ab : SUCCESS in 18m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a8bb941380964428aba52ebb34f6be76 : SUCCESS in 7m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5597d61d8a64a96819bc9eb4a048dbe : SUCCESS in 1h 04m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d328ba158d374907adb821499d29ed43 : SUCCESS in 1h 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b04118a5ef8f88c046eeb640186e894ed76b172e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 16:43:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfc415ed6c1e9273f993c31e0b769fe8cb876cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:43:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd7147e8aa60e914c4bf56a256ac5716c0a19658","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 17:34:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc1c4dc37602c2943a32de2a430fbd14aa980adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:58:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf71f6038854d2e9be3377a63d65f48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea09253846984ea8a18d2577ed971d18 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d004a58e23c4119b344699d5d8d2c99 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86a0e027e05744869667efe17be4e9e8 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e397a5102ba43b88f12957f4ba2ef64 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a9e01aaaf6f43c09e0371b24cfdc029 : SUCCESS in 3m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/69892d685ea54e6a9ebcc8eb84273ee1 : SUCCESS in 11m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6fa71e3617a145a983df0dbceb4e2e38 : SUCCESS in 7m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b4d00dd18f3d4cfeb48e418cc0236a4b : SUCCESS in 5m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/23aabee57291457c80808585fb9b744c : SUCCESS in 8m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4b05a5eaf59743d5a6ab7d3a4ed17ee8 : SUCCESS in 10m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4fbf85f81cb34c1abec062c139370fe7 : SUCCESS in 5m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9077192731c3452081d802f0370e472e : SUCCESS in 7m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cc6a50fc479349a5b49b7b8ac3cbe1a5 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c6d8c9fedfcf4eb791a407fa76e586bb : SUCCESS in 18m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f8c573c5f9f426c82a5185d05bd4a70 : SUCCESS in 6m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/edcfa3ca797b4ab59122d8197267054f : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/51f2009fc1864baaafd4fe66fd2451cc : SUCCESS in 12m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd958acda8164d8fa7351c7bae35f0dd : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/96550c91cf3c48b0a7d899e3aabef547 : SUCCESS in 6m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/accfd2c7dcb44fde8e4eef8a018e91a7 : SUCCESS in 37m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f2c425ef530646568ebfebabcdc5a084 : SUCCESS in 8m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ea8e7f3803cc4d52a68d373023a8d238 : SUCCESS in 10m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5b96e30af87f44049a988355ebdea4f7 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de61396dbb4449cfb4005b5c86cf927c : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86f1c50c7be645b8a7f928729cb04653 : SUCCESS in 30m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d4ad6bf59c4a4ddb88282637e86220b6 : SUCCESS in 7m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a21f074c8cde4dc29792f30955945f2c : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94f989c8be8a45918f84f6c2406cdffd : SUCCESS in 6m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/47c2b68d3c09470eb14dfe672bc24cbc : SUCCESS in 19m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3d0bfeed9ef1418fa18bfb89dfe73dd4 : SUCCESS in 7m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81194c1fc47f4456900198cf3b94979c : SUCCESS in 1h 07m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f41768c6f0fd4ccca015fbe7b973d942 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":1},{"id":"89b0a602c8f8d96ac253187704e5c62ccdd169e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:58:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f5059743c0eb832caa3fc2945bd260efc1d75d6","revisions":{"0f5059743c0eb832caa3fc2945bd260efc1d75d6":{"kind":"REWORK","_number":1,"created":"2026-05-05 13:05:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/987353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/987353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/987353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/987353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/987353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/987353/1"}}},"commit":{"parents":[{"commit":"05450163ad671a1fefbb3951117000850769416c","subject":"update constraint for horizon to new release 25.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05450163ad671a1fefbb3951117000850769416c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 12:55:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 12:55:36.000000000","tz":0},"subject":"update constraint for horizon to new release 25.7.3","message":"update constraint for horizon to new release 25.7.3\n\nmeta: version: 25.7.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Commit: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Change-Id: Iddf43b43928c538e2457d4aef0cc1b025829716e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I392f2a3e6d95bb670c3dc1bb5fc68feb2cbc87c1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f5059743c0eb832caa3fc2945bd260efc1d75d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f5059743c0eb832caa3fc2945bd260efc1d75d6"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
