)]}'
{"id":"openstack%2Frequirements~987370","triplet_id":"openstack%2Frequirements~master~I83bbeda68d55b2dd761a434cb10ab95e4de66a38","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I83bbeda68d55b2dd761a434cb10ab95e4de66a38","subject":"update constraint for osprofiler to new release 4.4.0","status":"MERGED","created":"2026-05-05 14:35:28.000000000","updated":"2026-05-05 18:03:49.000000000","submitted":"2026-05-05 18:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987370-new-release","meta_rev_id":"f70e273c1e03bfea8ace7b9509e5892e0a31cb66","_number":987370,"virtual_id_number":987370,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 18:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-05 16:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-05 17:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-05 16:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 16:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 16:44:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-05 17:34:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"0ce270e20e700aa1bac7368bf99fe72f9b09376d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-05 14:35:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d4e39160580ccd66c492c2c75a06b787c0e970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:02:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9dfa35eb724d5994a49b8308c5cfa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9aa8a2763af432fa5af8a2ce18865e1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9613ceddaa4c4d8798d1c83be0c736e3 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ec5241a4cde448eae54ba92f66b9e09 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d61a3c8c46a2494896cce6a0ac55a09c : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24c8de2947cd4c8e81e4c1236be23ed6 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19653fba3b8a4f3eb594b265b5a99d20 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82f8ce716ae14ee48274e9dbecb1bce3 : SUCCESS in 2m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/edade00e2a704492a1df9dbb8736c6c1 : SUCCESS in 5m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/76cba840cb10429c9403761ff45f3a28 : SUCCESS in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/387e7edb261b471a8aa8c886cc20fc79 : SUCCESS in 6m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4c2b1180cfd245159c8ddf5e1b0aadd6 : SUCCESS in 5m 15s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/181c1034bb4241e4b712873ff23a0e3a : SUCCESS in 3m 30s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a3c18f78e6f945988b4889110d7c2542 : SUCCESS in 7m 01s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/3c85ecc7f62b41d7a24f489da1052e1c : SUCCESS in 6m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d60f018b3716469b81ccccd8bb29e960 : SUCCESS in 10m 33s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/229ae805155b4fd283db2d7652c12711 : SUCCESS in 5m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2460021f75324ff499daa726a6fd14e9 : SUCCESS in 4m 55s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8adcab08c7f14bae8cabd23ff099748b : SUCCESS in 6m 59s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0c1a4eb16d4a4a0f931fb6cfb0eee504 : SUCCESS in 16m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dbdbf85065c4912a3c4e69c3c4acf20 : SUCCESS in 9m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a7a8dcebe890491592fc793264ad9b04 : SUCCESS in 7m 08s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a33552555284416fb80e11ae317f1c3b : SUCCESS in 10m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/628cd02c36e447e28cdd135a1435c866 : SUCCESS in 2m 10s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6685efb211054c52b927d337f9bdabcd : SUCCESS in 8m 02s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/598f1d03e91e42a89df9834dd5d264f0 : SUCCESS in 3m 59s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f365aacf4b1c4586bb12100fabe46bbe : SUCCESS in 3m 42s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/347aca375f274c9bb635a39d2d4da817 : SUCCESS in 10m 26s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/333b2abcff0a410dae6af3d80f31dc8d : SUCCESS in 16m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d80be5f652946a8ac79350bf36a766c : SUCCESS in 7m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/bca39c5df00c4ec3ad55ac24639731c0 : SUCCESS in 14m 04s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ff20bd6406d64e8088ead30e9209423d : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a324a1a480c04a49acaea0c3ed732644 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75f51dcddc8942a7bcc410e8bc906e71 : SUCCESS in 26m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0f179ba72c934148a529a9775f0f5717 : SUCCESS in 6m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bf94266357da42e793a85af35a55e1b9 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12fec092da804a30bfa19e415e7120e2 : SUCCESS in 6m 17s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1b1a35ed860c4f758ce6fc8e89334b29 : SUCCESS in 19m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e378759c9cb74f7a819dd052707b7bd3 : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9f380ba9d354cf2a90fa1551bf103cc : SUCCESS in 1h 08m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35d800dfd4a94a098046ecb824406d81 : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6293bea762a1a7ff34de1b7215220f2a46549a52","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 16:44:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e842c06e0bed54343f960fa9dc7c1d91067076fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:44:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bcc209446a8dd134731696d4240f3aad8aa8f8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 17:34:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"266ee58db35f4926e026867231bc38ec3ccaa12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:02:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/785dcef3c04e48c9b08b23fb5d8b7729\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70aaf760515b42bea6dca1fe1953f496 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67be34958db843cfa237c1979bca1d0a : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbebfeaf22cf47e2bff2eb62bf7e0652 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7569fc155c7a43c39b89010f8079e95a : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519a8691febd4663b2395e1b08828686 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/474e280890c74b0b86124055379b322a : SUCCESS in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3637dc063b434effac1198f921f7912f : SUCCESS in 6m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dfb59bf236224e5a8e21fcf7a8d6c743 : SUCCESS in 3m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d60c0e2a83ad4d28be8f97d78454c82e : SUCCESS in 6m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d8a9486794e5412c8801b5230fe155a9 : SUCCESS in 6m 12s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4034efe96c6a43e9b820fa6b65c587d2 : SUCCESS in 11m 17s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e01e2580975f462d8b76d4f663f84413 : SUCCESS in 5m 04s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2a26d02278584dc2bef17845962e9577 : SUCCESS in 5m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f41952804e104be9a2b4ccc955e85d70 : SUCCESS in 5m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a0bb6e162ece4d4d87a9d0d490f99234 : SUCCESS in 15m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c915f3f76d644818ab23ef3299e1c768 : SUCCESS in 7m 28s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a40ad00bca1848faae2ea6217b7e63f6 : SUCCESS in 9m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2a4d03a48a934cfd96dea8670462d2c2 : SUCCESS in 11m 07s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/64375361e6ba4f779cddab4898da2a79 : SUCCESS in 3m 34s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/dd59a1461b4e4b57b98b85450c318b0a : SUCCESS in 10m 00s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2177cb52bac2488f8d81807d358c7742 : SUCCESS in 25m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dffa456678da47f7a98baa7f6b2c6b8b : SUCCESS in 5m 41s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/afcbe2c4c40244aeb3f8d130721b89b5 : SUCCESS in 10m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/11d7143612c54c68bccf41f1a89dd401 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b0b131e55654e27a591e17e689ba3f1 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86bddcfb62904853b97351f208255686 : SUCCESS in 25m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/185631843c1848a38ef25a2a0b2f61bc : SUCCESS in 4m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/16b48bbc9fb047bf993268ddf51fa21b : SUCCESS in 3m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5248def36634894a4911e3f643a288e : SUCCESS in 3m 59s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d448196217f944649501dc6a8a21d7d1 : SUCCESS in 12m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b80ee56da4a7486d982d0cd75240240b : SUCCESS in 6m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43409c30a4b845e58b52aff63892365a : SUCCESS in 1h 07m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d9d3f60b65b49cfb8b974064e2eeb22 : SUCCESS in 1h 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a9abc5ee3a4393b00c9417b7464a54ee298e9014","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:02:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f70e273c1e03bfea8ace7b9509e5892e0a31cb66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a14d5a46fa94efdaf3f05b8e792a268\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48696983a0c946eab5204542081ba992 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d868f6f022cb70b84f0f6762226712d3ba2785d","revisions":{"7d868f6f022cb70b84f0f6762226712d3ba2785d":{"kind":"REWORK","_number":1,"created":"2026-05-05 14:35:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/987370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/987370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/987370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/987370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/987370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/987370/1"}}},"commit":{"parents":[{"commit":"ed679515f6e053d20b05ffcc30b3172de864f7f1","subject":"update constraint for oslo.privsep to new release 3.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed679515f6e053d20b05ffcc30b3172de864f7f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 14:25:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-05 14:25:27.000000000","tz":0},"subject":"update constraint for osprofiler to new release 4.4.0","message":"update constraint for osprofiler to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I55b7dc48e7d9d963bbf35868a23901bc2d6c051f\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I83bbeda68d55b2dd761a434cb10ab95e4de66a38\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d868f6f022cb70b84f0f6762226712d3ba2785d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d868f6f022cb70b84f0f6762226712d3ba2785d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
