)]}'
{"id":"openstack%2Frequirements~987973","triplet_id":"openstack%2Frequirements~master~Iac60c9c770b5dde384906b7fdfa97110d70d56fd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-05-13 09:13:44.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Iac60c9c770b5dde384906b7fdfa97110d70d56fd","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-05-09 08:16:55.000000000","updated":"2026-05-13 10:46:57.000000000","submitted":"2026-05-13 10:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987973-openstack/requirements/constraints","meta_rev_id":"2c152445fe150ed2a5b68aeb30cbae00cd6a860b","_number":987973,"virtual_id_number":987973,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 10:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-13 09:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-13 06:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-13 09:13:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-09 10:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-10 19:39:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-13 06:11:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4012b152a4b2a365f195c0aa267f40158a10a222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-09 08:16:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74dfe8ecded8e3ac7e80ea3c3e10b40066cadaab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 10:17:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15b2481ce56b4975a769712b84171b2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a884f50a3648789cadee870cb9b0aa : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94ffe8fc03e4e539c0ee924003d6025 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcd0cbe33b6743c5922a2d2ae72067e7 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79da4db40c1a4d64947719880c536649 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f7bf816db57b4ca891fd824c9a01b5ab : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43cc6d57d3c64fc1b7218c14b1e047ce : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0529a23d3fc4a99ba95ba25831b3ea9 : SUCCESS in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/532f5923146b42c4862b45c0df512305 : SUCCESS in 5m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/88af687532394a56ac613cf5b0c107e7 : FAILURE in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/50f465e0eb8848fa85d43aa2a6eb9b3f : FAILURE in 3m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1138337ce7dc46a39e7813a3c2e7bd4c : FAILURE in 4m 00s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b3d73779e1db40c9a1bcb44d1d800771 : SUCCESS in 4m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c151c8f585ca44768041f0350bc64ee2 : SUCCESS in 4m 34s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/830c32b2b98d4fb0a07d7556b33632aa : SUCCESS in 4m 36s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/7bece678bb16444c8271f5e862d0a891 : SUCCESS in 9m 47s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/30dd0f5fe8c44e7f9e4809631f12ed3a : SUCCESS in 4m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5a520a777636412db674d8bc4e4db9c2 : SUCCESS in 4m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3c33dc906b8d4b5eae4d74aa8f113369 : SUCCESS in 6m 07s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/75468cd2579d494499a0eb0ab2398e2c : SUCCESS in 15m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/916c893c88af4f81ac67820440f7e140 : SUCCESS in 6m 25s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b1177c14476d4aa4bbf624371949fd12 : SUCCESS in 7m 44s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e3e2f7c93aa84b56a6c7df041ab9efea : SUCCESS in 12m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/841c997dce1944dfb02791b71a082524 : SUCCESS in 2m 25s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d8ac04fb0ae049099010c841534079d3 : SUCCESS in 8m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c0c540577ce34dfea93e6e0e1f76cdad : SUCCESS in 4m 31s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/05d46d4a65f441dab9f7f856aa50a15d : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/9d39e7d714294a13aa6f1288c72fbe12 : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3bbd18141ed948a195ca30f23dbbe187 : SUCCESS in 19m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e70ff348dcb4e5eb80959b6fbc294d0 : SUCCESS in 5m 49s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4c44170239ef479ea620f395f2979f72 : SUCCESS in 13m 39s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/fb97f2910d204c1c9913ab6bbb730321 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b9a8ffe60d748179ca2129d1ba5afe8 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9f761deff5e450bb0c2d6f9d090722e : SUCCESS in 28m 25s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/92abaa8c169340198c952a16694684e1 : SUCCESS in 4m 51s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/17002518ada64c0e8e42ceb755afcf9f : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8ed3bb0fd4648119d3845235ba4c4de : SUCCESS in 5m 27s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f82b31d6336c4e5596d3e63ebe86b21c : SUCCESS in 18m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d2d188562af74f3da896f41e4036cab9 : SUCCESS in 4m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36ba1b0b62044603bd67004a61d1b98a : SUCCESS in 1h 08m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcbe5c47567841309e0d3a900f6fc03b : SUCCESS in 1h 59m 02s","accounts_in_message":[],"_revision_number":1},{"id":"42490c5ad124bcc55ca8b1c80ae84d8e1220bb63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-10 17:21:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f7c59d5f4de83080a9d8ee1e10a3cb5dcf4654ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 18:25:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765ca5609cb047cb96cac7f076dae7ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84512a0570fb4459b9939cc08d69e299 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ef399e20e04be8895f24ca3a304142 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8e31158dffe414db2a10d8e660450af : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2235dd5f5901464dbb3b65601b4adc73 : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/326242d1a4e04001b7e7473893fef9bb : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c12d32981c644a48a510d136fc6e16f5 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45cc0a4833314ee9a737aa531234bafd : SUCCESS in 2m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a86c3b1c756a45ea91846791922b89a5 : SUCCESS in 4m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/87e0071bf3054199987bd6f4688ad08d : SUCCESS in 6m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0db7b0325bf04635b7f046693c6f2a24 : SUCCESS in 7m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/00a9d71b57e64e9c8d1d4d7a75b1dbe2 : SUCCESS in 9m 29s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/24238a46683748ecaabbd1bbf98d866a : SUCCESS in 4m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/247381996e284fc9b50cd34a125d8b48 : SUCCESS in 7m 11s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b9d969fe7a6345dbbab1f73a8cda16f5 : SUCCESS in 7m 12s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d35476cb4dfe45d48854cfed0a2acfa6 : SUCCESS in 8m 04s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1c07e835a4a840ac903f653ca1e18f8c : SUCCESS in 4m 29s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/37b8b80afe9548dca5c680993d1e1410 : SUCCESS in 5m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/054da24e07c8449b940466dbd58f693c : SUCCESS in 5m 50s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5a4ea68b70bd4f40b455e0b967220113 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d1781907b5c4e27969671e5b2bb214f : SUCCESS in 6m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/59412b45bf814b42a0d72e4a82773323 : SUCCESS in 6m 53s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ebefcdbeefed4478a22c9809d326a803 : SUCCESS in 11m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7d9721586a814c8395b8736cdbc10625 : SUCCESS in 2m 45s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d3783e2f3367444ab2878b347cfff106 : SUCCESS in 8m 58s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f6df9ee54be046f184d16741059843db : SUCCESS in 2m 18s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/bc074b9d97da4a7da0ab5cbfb8d4a2a0 : SUCCESS in 3m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/8f3730f72cd646e1a67f6b3f975ba16f : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/38cae2bcb90c4eb387dbedf231dda33f : SUCCESS in 15m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/788e68034f274e80a84f1a2b2c42e925 : SUCCESS in 5m 56s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/52666b0cf87641378a9f6191e0a8168c : SUCCESS in 10m 11s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/795d952ffdb24bdfbfbc47fdd29a118a : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3897168bfeaf4d9689d80c854b6b6770 : SUCCESS in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e785628e37b4aa8a81f5aa6c419f115 : SUCCESS in 24m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/07c8a43a9bc4405ab3ed9edda11fc5cf : SUCCESS in 4m 33s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/45791ee5f9ad48de9a747683dbb1309e : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebd26d3771fb403caaedc855395369c3 : SUCCESS in 4m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7a78d539f1c740268e72d071b54957df : SUCCESS in 17m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0e5562e71d684d96b7d97f65523c29bd : SUCCESS in 3m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a680e564be44ef8bf1e58bea67b2f73 : SUCCESS in 44m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/733f307eeae24548996fb75fea5e7445 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":2},{"id":"24a9169772f3cd09ffe40e688eda65e65aa12499","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-10 19:39:51.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89ede65907e83f0482762e94096dc9c391b1443a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-11 07:34:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"31e8798bb1134729884e18c8f3555379eef9fde1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-11 08:13:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"35ff47859f0bb7999fc6c888367186d35e6573a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 10:02:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/074445c8eab14543a46670d4328b564a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af12c42da4414b5c98f8a83b40bacf65 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3ab1272fe2411991e2f71080869221 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/047058e6736d49d6a3ab67c22099f7dd : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ecd42ce8d174dd99d797ecd1fe30755 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/325ab315523d48dab2e127c30583615b : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/512f8b2770e74ee293f94fb46226dfab : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3f34142261d4ef3aa2560e065785616 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20b5dced0613430484aa5889f7a8f4f1 : FAILURE in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/99027b951a454410a19d13e618c1c5b7 : SUCCESS in 5m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec4202fd5d2d4e3c9173571ff7928fc6 : SUCCESS in 5m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e4c546d113747a79ab7d1f3233bd89a : SUCCESS in 5m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/17eb2bc3f8ff463b8c2c78fa240940b3 : SUCCESS in 4m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/69e6ba910f894a5089eab3504db0564d : SUCCESS in 4m 40s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ceff927ad4ad4e329ae223e7c0d1d8bd : SUCCESS in 4m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/bfc0f4b5bf2742bb9f013c153fe5abca : SUCCESS in 10m 40s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f0306efbf7d14a9bb5d5e4662290af5a : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/49c2bdea02274d228d1e2113bb0519f6 : SUCCESS in 4m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e084a863db474120a6acc29f97c900d4 : SUCCESS in 6m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2103be81299944ec82bd7a1471dbbe50 : SUCCESS in 15m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62cfb7c2c3794954ae580d2974669555 : SUCCESS in 6m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c18852846e80442a966456e5ffa47f0d : SUCCESS in 9m 32s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/494f7e1a3e994c809f05fecb929a3f0f : SUCCESS in 11m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0a9407c62bab4a22aeac309d4ecd8dca : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c46f09c8bf474ca9907603ed76db3106 : SUCCESS in 8m 43s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/321bd372c4304f33b2ea694ddabdad31 : SUCCESS in 4m 22s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0767ffec002c4a4593dcc05b694f8619 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f1e436e62baa4686be7c8c99025527ea : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7db50e3858114c8ba6268c0281f3ae84 : SUCCESS in 24m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4436186f79154b4fa5809558857486df : SUCCESS in 6m 00s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a0297e09265a4ff7bc1ada536310e196 : SUCCESS in 14m 59s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/175a3e88b45349f5a3dd127bee12f554 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eeb6ed8c007940f3913157af2c596c60 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b36d6d19c0214816b4594784463807ea : SUCCESS in 24m 41s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/241415a8fa6c455a9e94a8f5ce101f49 : SUCCESS in 5m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/340bbc7c790a462ab75d71aad6a34a54 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d118afb870ff4dc6af4a9b42fa3f8902 : SUCCESS in 6m 41s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6ac0f26db2134acd8255ad4ad21b6040 : SUCCESS in 17m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7a9130b0e6f7495084bf70208d414920 : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c0e5ac841ed4c4cafd99dd9750d2843 : SUCCESS in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20ab12d764bd4505bf4226fb8820d0c0 : SUCCESS in 1h 43m 05s","accounts_in_message":[],"_revision_number":4},{"id":"95da367c2d53de65006d2f6ca026f4d69569d21a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-11 15:43:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":5},{"id":"35bad795e6c3b26b2fcc4e3a54d9828c088f2bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:32:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/836ebe8ddf384816949f56acd6211906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4c440919ce4cb6946a6fd1d11c75fb : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5c89f02283426396791d47920dbaaf : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67385fc54647417097332237e9ba0d10 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ce8bf06a4f146478d0e5a23263dd187 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e2c44ba767f3483eb914726fe09a9074 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee266f118c0745ca8cff36a0c71c7200 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3de84aaf3d9a4c00b2c4b311c5bcd356 : SUCCESS in 3m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fffd1bc65bc346dd99e15ecaab76b602 : FAILURE in 3m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/791832d7c82d4d7bbb1b0098e7e66b53 : FAILURE in 3m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/824c4b3e93ea45a786692e6cbe86b7f7 : FAILURE in 3m 32s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/803b147fd46b43939549b05f162cef05 : SUCCESS in 5m 44s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/dd575df4742a48a796eec6658d317aa1 : SUCCESS in 5m 38s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8a0be7f9a3fd48afb72400924075ba34 : SUCCESS in 5m 25s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b23b03a9b0c647baafa00f3b5bd9c49d : SUCCESS in 8m 22s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2ae3b8da05ba4e6a89fc5f155fe23826 : SUCCESS in 4m 47s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/12d7930d57a74987a14e60304e5f355b : SUCCESS in 2m 52s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/39121494eb254ac7a6208b10144d5bd3 : SUCCESS in 6m 23s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c50750283f73476883ecfd60595b4f65 : SUCCESS in 17m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40a9c5a5bdee4a1ca50a1b474747376c : SUCCESS in 4m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/060e592e3b4b47519244e654a54be7a7 : SUCCESS in 9m 00s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8dbe499e8b6145ed935b43771cca49ab : SUCCESS in 6m 50s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ee4de8d374f84fe49023f1ad23093842 : SUCCESS in 2m 03s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ae08f9c43dcd4f3cbeb6b017c822f278 : SUCCESS in 9m 51s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9fec6046566549e3a19b646c7e0b5eeb : SUCCESS in 4m 27s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7fe91183050546f3b788e20663d75359 : SUCCESS in 6m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2fb48bf6f4254f5da698d790b035c95e : FAILURE in 10m 33s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cc2decb7467f4f2e8913e05a6e0c52ac : SUCCESS in 25m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f395183584674ab6aecd3d2c6e617d4f : SUCCESS in 6m 49s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4be8f1da0f94444592404f2043df7b50 : SUCCESS in 14m 27s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/35422daa10554cd48cc653963bd26ac0 : SUCCESS in 3m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39f5bdbddbc646f1b6a59d8aeef2fded : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f802758289ce4586b0b3bcfa8c39df2e : SUCCESS in 23m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e71d48da54064f2ab3229ad341c5d903 : SUCCESS in 7m 00s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a8d9c19d7e7c424483f05351a8b13fa5 : SUCCESS in 3m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e2106f0d0764296802d873f26363164 : SUCCESS in 7m 37s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/993aafdda1f648159677f815c7479d7b : SUCCESS in 18m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/40a7fe90db86493cbe1f98558b2c6667 : SUCCESS in 3m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ed6825e633347b993affa990646b3af : SUCCESS in 1h 06m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6105795523d44e8080bb94daceb6d136 : SUCCESS in 1h 43m 05s","accounts_in_message":[],"_revision_number":5},{"id":"bf6e4c906da967a7a8676b82c05c36524c974c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 07:22:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":6},{"id":"190f3aecb427976e05f2e160639c224bbc3d09f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:20:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfa070a0825240ac85e87d8b246c1271\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c532557486347759fe4ea6ad2c36bd9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ef63c3f1794563955a5e62e9ed0d7b : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5716c7623b1f4e63ae09cebae112122d : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36d653cb8df942099cee100bbd8d4a4b : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56b6eedbd7f44f98b79a4439e2c862dd : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1de613cab2774a29b21fa4e599d54414 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f9387d7e97447e1976a7d043c0ffd38 : SUCCESS in 1m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/20fb8c71be994c6888ff20b0a1f61498 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7625813e6c8b468abbd432fe731501e7 : SUCCESS in 9m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8a8a6ad5f6d24a5fb0db8c20cd3ffab5 : SUCCESS in 5m 28s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/273fde84b703495895876289bd8c8ac0 : SUCCESS in 6m 22s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f892103f01724f68bea0b119c0953fc1 : SUCCESS in 4m 21s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1e98b589e39147508b0c9e351aa3fecb : SUCCESS in 4m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/471a32ee100d4d50875934c623fc159c : SUCCESS in 8m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b17c11c8274749d98a77dbca2526562a : SUCCESS in 4m 36s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6787ee361bd3434889b20f20fb513738 : SUCCESS in 5m 06s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d5724a0f102a4e7aa1d275f7cac8668a : SUCCESS in 4m 07s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6fe95bdf24e84860be0c290ea9abb6cc : SUCCESS in 21m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24c9040df7644d3d8e84b5bbca2a668c : SUCCESS in 4m 48s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b41214be11cb41b68227bb81103d207b : SUCCESS in 7m 41s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/090e18738fdd4c31b620d4f727950294 : SUCCESS in 6m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/32f3a190f1ae4152bcf3983383de96b8 : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/285f911a26394e12bb8a6c13f13170ca : SUCCESS in 7m 23s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/3f2335816b5047ca96bf256963734a2f : SUCCESS in 2m 29s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/3809f525792c4d4e96a4f4adabd48315 : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f0fcb8d5abd84d4cb80d9a11fed53170 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/64dbc25b4a0c47b287ad8f02c9fa0988 : SUCCESS in 19m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3cf5ca4417e4c67879e328076722d24 : SUCCESS in 5m 50s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2b11f605369940a999789ca80eccc158 : SUCCESS in 10m 15s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/41815f5d60134079a418f7f6c8647634 : SUCCESS in 2m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfe251e2ff5742ddac68622076444cf9 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b2e5f5268f840e2a21452b6e577e3dd : SUCCESS in 25m 51s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c4bb281226f94d22aba2223ab1a20377 : SUCCESS in 5m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b608e3e6b58a4cb1b3ee919e66156d2a : SUCCESS in 3m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46c68aa52de34307ae44f65cf3f2b00b : SUCCESS in 8m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/980fd17e8d034e30902ca65520e6b3e3 : SUCCESS in 11m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1696f4f18f614562adcd7fa69d602f34 : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a99844737423409b9b3bb8a2a1e53a08 : SUCCESS in 1h 01m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbf03273c5654507b317e767564df1e5 : SUCCESS in 1h 47m 49s","accounts_in_message":[],"_revision_number":6},{"id":"39a2f7519c5b3d5b0200878c4afe426fb371e731","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 09:38:27.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"ea97e8f67a0ee225ddb1675d85fb11654a6f5783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 11:16:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"7e42f5d0ed83435941b3c8ec8cf073937103009f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:19:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca343a6a029455791fc6069dc45d697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a6f1e9cdf54cc3813649af872b0895 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c9ec01aa62460d8ec3786f52707428 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5992771f1c3f4cb1b4cb7fef18c5d0c9 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/295170eae89a4aa5bdbccb95dc9c89f2 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd85c7ced14248bfa5d0e3b0516288f2 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dd01a8c6b084b488eada1088345a14a : SUCCESS in 2m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3171539c53641fb91ae645fe49bcea7 : SUCCESS in 3m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/293fffc463d74349b0830c15ada69355 : SUCCESS in 5m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7c9dd68de80c4c3b98f588a3957ca41f : SUCCESS in 6m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/debb0b49930045a2ab1c1d621a7aacdb : SUCCESS in 5m 31s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ea66b4d0ecff4ee4885bf95a2c40a34b : SUCCESS in 4m 59s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/31ac9250c8944e5eb99d4c02c9a9bd2f : SUCCESS in 5m 32s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8ff6398d9aa44476b1d92d52d32d615d : SUCCESS in 6m 20s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d1eca2b25724440ab5f1186585c55c96 : SUCCESS in 10m 01s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c0311afba2654bb18e316f984ae3bedd : SUCCESS in 4m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/86da340c265f4c918415cb56481911a6 : SUCCESS in 4m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/91a676041fb54ac8aa08b741480dffd9 : SUCCESS in 5m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/05a7e738882b41559198542a5ba8631a : SUCCESS in 17m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1aac2e319bec4f79a71599ca0d680575 : SUCCESS in 6m 21s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/cd55fe6535924820905cad069eccfd90 : SUCCESS in 10m 22s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8eb37c8ff47c4c5bbc8820917f983cea : SUCCESS in 10m 40s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/4415a1f76e454364b4eab32fd6c3157b : SUCCESS in 2m 28s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/dbaa2504a72b4c8399d17cd5910d5659 : SUCCESS in 8m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d21a991b5dd14806bcb0a8a881fd0ba9 : SUCCESS in 4m 29s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/78155756da6e4d2190b81373105a873f : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7ace08d4f82147399982c9e680d4340b : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/e72bc4aca70f445290e63872d0b654db : SUCCESS in 19m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1ae34da55e5d488fbbc4d4f0b51c783b : SUCCESS in 4m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6a466dacd72749938146c568dd7ca919 : SUCCESS in 13m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/94d507562fe44ae5989bf0e5b07452c1 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a7c73bd25c34517ba70ee7b2ae2e131 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fad2a03ece044cbeacf553f985f9a4c5 : SUCCESS in 21m 05s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/546cf04f867145b1b10ba40a8ace6dea : SUCCESS in 6m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/68cd4bfcef5c4190b77a942e7d059587 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e5f6d4a959a4768924f0d2ad46e8de5 : SUCCESS in 6m 47s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/24459ac5ff38468a9d03790046f3b49f : SUCCESS in 13m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8a2617459a134b81b195350723421bd4 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5f0c7413b4949afb45023567c00a0b1 : SUCCESS in 53m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cdcb7bb2ffe437d92c2afdc3e1bcf00 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":7},{"id":"1a0ba24678a3c5002fcc6b58e4a2e3430835062a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 06:11:50.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e1a7e4bc383634f6e45b62ad77e082d0e714e5fd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-13 09:13:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f71d82c2a2359bc474f4ad0a049ae25f43786a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:14:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2ef98be9f81c8a82b38c0e2a78db28dca8ea4320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:45:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf722f75db04db88018bc6cdac73271\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943889210d0c4b6d8f94e7ed7757ba62 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03118f37eb754f3699a6c23490b4479c : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d174bc78f0134033bd0da30bff43e69a : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09f542f753ae49e38a185388f8baa623 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d7a699758b94ebf90f27f898966d812 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b2a22f9b4b54ee5ac8ea7e27d39f630 : SUCCESS in 3m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b443341e75bd431ebc081aa421e9e8a9 : SUCCESS in 3m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/358462f005ca44c9a26612f303f4137e : SUCCESS in 6m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c20773d66afd4a82b8ee46b51f256845 : SUCCESS in 5m 53s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6b6c8e31630c484a87dcdd3c84d87abd : SUCCESS in 9m 48s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/66dae52a20994c65b59df2ebf86a2236 : SUCCESS in 4m 18s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9a61b377429849a79ffc0887082c9d83 : SUCCESS in 5m 21s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7156addc06fb4f658b38eeb8d1cc164c : SUCCESS in 6m 17s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ca12ffbd1b224fdfb9dc0555e7fbe956 : SUCCESS in 14m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c558069215ee428dbf29f23a2d895b19 : SUCCESS in 7m 05s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/746e1056a9c2418897bc1035a80b6c41 : SUCCESS in 8m 06s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2c0745d87666405dbacf6ac739ae0722 : SUCCESS in 11m 23s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/da6e4978f8904d898554b909b7601c45 : SUCCESS in 3m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ae708e9e95b046bc85dee1f63e3a7472 : SUCCESS in 7m 57s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f430d9b8c8c5463ebb402e3900eac778 : SUCCESS in 21m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f2228f34ee5c4729b05e5fa9e2357116 : SUCCESS in 9m 22s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fe4e08aafc73464b933d750f5891dc11 : SUCCESS in 12m 31s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/eb3bfbf44ceb4bf1b3f2c3ffae36256a : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edac0d85ea32435493f9c1492e443067 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/203acea984f24045a4a973e55b83e332 : SUCCESS in 27m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c1cebc3938e74e9a91fdfb4ffc02d484 : SUCCESS in 4m 57s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f2d2886f7aed4380bc820d62790f2b7c : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0563ece618334d46980657be1cd2d9a5 : SUCCESS in 3m 40s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a266229a498d46e19ab088272ef59c85 : SUCCESS in 19m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/455576fa796f43099bcde8e03a24f7bf : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbc10619c6e2408ca80532891234d0c5 : SUCCESS in 33m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/331b9efd3baa44c9ae0e1dd598ef96cd : SUCCESS in 1h 22m 24s","accounts_in_message":[],"_revision_number":7},{"id":"e71ad047f2c5912a7f2489ce38555dbc39af2ff6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:45:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"2c152445fe150ed2a5b68aeb30cbae00cd6a860b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:46:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a778d4388c4bc283e15249fc55a951\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd69c2efe1540e6a865883aa3fd69d3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7546f1127d2e172d0cb8c92076ef365b320c6bac","revisions":{"fa9263900015f3bed8a1c78f6c2aa175bd25ce75":{"kind":"REWORK","_number":1,"created":"2026-05-09 08:16:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/987973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/1"}}},"commit":{"parents":[{"commit":"31265f64bbe899a51e6a921ca67031310e864c42","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31265f64bbe899a51e6a921ca67031310e864c42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-09 08:16:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-09 08:16:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9263900015f3bed8a1c78f6c2aa175bd25ce75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9263900015f3bed8a1c78f6c2aa175bd25ce75"}]},"branch":"refs/heads/master"},"058ac4b7b7bc6fe0a8a7769f95e5c6615ac20989":{"kind":"REWORK","_number":2,"created":"2026-05-10 17:21:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/987973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/2"}}},"commit":{"parents":[{"commit":"31265f64bbe899a51e6a921ca67031310e864c42","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31265f64bbe899a51e6a921ca67031310e864c42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-09 08:16:54.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-10 17:19:21.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/058ac4b7b7bc6fe0a8a7769f95e5c6615ac20989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/058ac4b7b7bc6fe0a8a7769f95e5c6615ac20989"}]},"branch":"refs/heads/master"},"7511f5f56614bbbdc7b7206bdc9fa1ef140a8051":{"kind":"REWORK","_number":3,"created":"2026-05-11 07:34:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/987973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/3"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-09 08:16:54.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-11 07:32:53.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7511f5f56614bbbdc7b7206bdc9fa1ef140a8051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7511f5f56614bbbdc7b7206bdc9fa1ef140a8051"}]},"branch":"refs/heads/master"},"ad4f1ef216bec743f5274c66940afc92ae2f267c":{"kind":"REWORK","_number":4,"created":"2026-05-11 08:13:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/987973/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/4"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 08:13:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 08:13:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad4f1ef216bec743f5274c66940afc92ae2f267c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad4f1ef216bec743f5274c66940afc92ae2f267c"}]},"branch":"refs/heads/master"},"86567bb8442151972e10bd3460a32cebc676654d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-11 15:43:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/987973/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/5"}}},"commit":{"parents":[{"commit":"ca1bddbab6dd0334018dfd0d7759fb53309ba595","subject":"Drop support for python 3.9 and 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1bddbab6dd0334018dfd0d7759fb53309ba595"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 08:13:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 15:43:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86567bb8442151972e10bd3460a32cebc676654d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86567bb8442151972e10bd3460a32cebc676654d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ad4f1ef216bec743f5274c66940afc92ae2f267c","theirs":"ca1bddbab6dd0334018dfd0d7759fb53309ba595","contains_conflicts":false}},"7e7e00dd39cac748838feb6d04269d4e987453e6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-12 07:22:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/987973/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/6"}}},"commit":{"parents":[{"commit":"23cdc1c30ea90acc80827f8f2684b97d27fb179f","subject":"Drop support for python 3.9 and 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23cdc1c30ea90acc80827f8f2684b97d27fb179f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 08:13:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-12 07:22:06.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e7e00dd39cac748838feb6d04269d4e987453e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e7e00dd39cac748838feb6d04269d4e987453e6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"86567bb8442151972e10bd3460a32cebc676654d","theirs":"23cdc1c30ea90acc80827f8f2684b97d27fb179f","contains_conflicts":false}},"7546f1127d2e172d0cb8c92076ef365b320c6bac":{"kind":"REWORK","_number":7,"created":"2026-05-12 11:16:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/987973/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/987973/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/987973/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/987973/7"}}},"commit":{"parents":[{"commit":"23cdc1c30ea90acc80827f8f2684b97d27fb179f","subject":"Drop support for python 3.9 and 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23cdc1c30ea90acc80827f8f2684b97d27fb179f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-11 08:13:22.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-12 11:14:38.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Iac60c9c770b5dde384906b7fdfa97110d70d56fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7546f1127d2e172d0cb8c92076ef365b320c6bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7546f1127d2e172d0cb8c92076ef365b320c6bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
