)]}'
{"id":"openstack%2Frequirements~988048","triplet_id":"openstack%2Frequirements~master~I5ee2ba55316fbad5dab6d63131950b3242df9314","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-05-12 07:23:04.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-12 11:08:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ee2ba55316fbad5dab6d63131950b3242df9314","subject":"Drop support for python 3.9 and 3.10","status":"MERGED","created":"2026-05-11 08:19:23.000000000","updated":"2026-05-12 11:11:02.000000000","submitted":"2026-05-12 11:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988048","meta_rev_id":"6178967b72fce3e924506dc0d335b3a571cdb243","_number":988048,"virtual_id_number":988048,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 11:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-12 09:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-12 07:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-12 09:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-11 09:58:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 16:03:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"676eff3f5af46bc70b20cfa35f868012cea844bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-11 08:19:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8a681fa1df0f1a57f2a65f26841d4b51db86a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 09:58:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985b6329eeb54a4e8c22c9e3b799a889\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/320cdd9af76148d4a72f5758e5dba334 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3364f8de364e405d9a57eff9fc3bf29d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2bc6c2833d4c1892541c73d1518790 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/136c5944ca9d4a7382f313d5f6d5372a : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eaf14f838d1e4449a6e2e9b86d0f09c9 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/813e490e99c84b1bb8dc7ba027b58ece : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adfe4523641b4317b50287132b7056c0 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/309ed1a7e2fd470c93ea5e6abc983f34 : SUCCESS in 2m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13fd4d5391db41bc8a33cde3bd256df2 : SUCCESS in 52m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7535c4ccabb4fbd91c749452a98ca24 : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c4c55c71b8ac6c13ce773d31ae6126cde1c61e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-11 15:42:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e891a8431cdd97c915c0f3796c2c805829278366","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-11 16:03:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25faf297905f23adea75f8a56f1e6381f02b4b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:06:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2f8bb125444784b9cff6875a4cbab9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0642f22ccc804dbdbd6efbcd8e6aabb0 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709df8fac50d4c168ac312d70a295580 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa34149368f44a59852d44c95a9f8f7 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b4ebe3442d74728bf1cf90b060371eb : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/929c4ace331040e9a8ae02571840eb32 : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07c89b23b60440498054571f9c96303f : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bf238722d964ce89cad84f46bc8f450 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94eda0538c1a45d384c5b29596df4d3f : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b255c3c7bd0d451ca67977dfebe15a6d : SUCCESS in 54m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f1c9c4e296e4ab3b6daa8ddf79807b1 : SUCCESS in 1h 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"04733233b32b249612fb5d97e12dff879abc1d14","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 06:43:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fc6dbd376eddfeba1e52548482f0abf5aebd4b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 06:46:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c6ba097d50ab8c4339c99a48fe52b5668e04f53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 06:55:34.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"35d44f2863d9797669eda279d18672637905ca93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 06:59:51.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a75c6c358a0e9bd3ef9a3e281de22d5b36938cf8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 07:09:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"502364921a7eca95667496407c9ef1135d4b8f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 08:44:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3428c9b546694aaf9bb13f77b2d85a9b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26dcbcc5e5794f358f53bdcb49ef9146 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79077eeccd1945ceb1884e21531a0b44 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdb5a4036de4913b790aaed9d19a650 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e925b6f4887f47eaae7e05d46d018287 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0532acfb5c024fcf818348e9f206ac60 : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f290c62cf5b249d8922736b19bef0c2b : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd7822c288d34f4991f1f74805bb8a42 : SUCCESS in 2m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6d682bfc74b4b0e9f8da4ab1a7db1b3 : SUCCESS in 2m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7a7f5a77af3444cabcb8ec6a91825103 : SUCCESS in 5m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0e04be1663334006aba39c4014cfb897 : SUCCESS in 3m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/25fd7c7540bd43ec8c3c542163fa066e : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/102c47ad43064172a5c04b82379ac863 : SUCCESS in 38m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b83845e5bb4d487d955da84e6c43d8d8 : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":5},{"id":"2058e3a563fe273ac1116b73b9c7257bf5f781c6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 09:36:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bfb70464d506fe17f6a5442c3a8bf6426b816473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:37:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"56b9c729d5389a855e54e30bb495a894537d969e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:08:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/774d3609ca0c44fdb4aad5ac72c7bdb5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d3294e7af3b4b3789d762552d314607 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e96e9ca0cc458db4b3874a4561c04b : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b9709f95414f74889c849edf443787 : SUCCESS in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b66f679c30934e77841e42e0f6ac5c99 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42b4f8345173428a9319d0e5dbdafceb : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6427f23aa5f4d1f8b6e97e7d9e72701 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cff088a9be5a4dbfba9912bb00bf7773 : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/47198413aa6a4fefb1b9ba0bb96d1919 : SUCCESS in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d690b27ff60443ca6d50f4178d3ed73 : SUCCESS in 6m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e8cf5c0eeada4b36ada1591454770646 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc0a80052e2947bdba5c897abb02015d : SUCCESS in 1h 22m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43ecdbe5cb254ad18a653b60593887df : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":5},{"id":"741b6803f35000c1b2290d867a1a3b22caa8db66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:08:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6178967b72fce3e924506dc0d335b3a571cdb243","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:11:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d1d2773ae048208025990351d48524\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa924e557fcf48358ec1bbb955916669 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"23cdc1c30ea90acc80827f8f2684b97d27fb179f","revisions":{"39ae09b6887b8d2e3e02ee457d2513c00d225e43":{"kind":"REWORK","_number":1,"created":"2026-05-11 08:19:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/988048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/988048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/988048/1"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:18:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:19:21.000000000","tz":120},"subject":"Drop support for python 3.9 and 3.10","message":"Drop support for python 3.9 and 3.10\n\nHibiscus (2026.2) PTI points to 3.11 as lowest version [1].\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I5ee2ba55316fbad5dab6d63131950b3242df9314\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39ae09b6887b8d2e3e02ee457d2513c00d225e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39ae09b6887b8d2e3e02ee457d2513c00d225e43"}]},"branch":"refs/heads/master"},"ca1bddbab6dd0334018dfd0d7759fb53309ba595":{"kind":"REWORK","_number":2,"created":"2026-05-11 15:42:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/988048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/988048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/988048/2"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:18:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 15:42:48.000000000","tz":0},"subject":"Drop support for python 3.9 and 3.10","message":"Drop support for python 3.9 and 3.10\n\nHibiscus (2026.2) PTI points to 3.11 as lowest version [1].\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I5ee2ba55316fbad5dab6d63131950b3242df9314\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1bddbab6dd0334018dfd0d7759fb53309ba595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1bddbab6dd0334018dfd0d7759fb53309ba595"}]},"branch":"refs/heads/master"},"3076cb54e229c199f89b6c25284fc54d0bfcefc4":{"kind":"REWORK","_number":3,"created":"2026-05-12 06:46:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/48/988048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/988048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/988048/3"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:18:32.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-12 06:45:54.000000000","tz":120},"subject":"Drop support for python 3.9 and 3.10","message":"Drop support for python 3.9 and 3.10\n\nHibiscus (2026.2) PTI points to 3.11 as lowest version [1].\n\nMake sure relevant jobs are triggered by tox.ini changes.\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I5ee2ba55316fbad5dab6d63131950b3242df9314\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3076cb54e229c199f89b6c25284fc54d0bfcefc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3076cb54e229c199f89b6c25284fc54d0bfcefc4"}]},"branch":"refs/heads/master"},"573d4ccdd553971d268532beb807d8ee17bf25b1":{"kind":"REWORK","_number":4,"created":"2026-05-12 06:55:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/988048/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/988048/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/988048/4"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:18:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-12 06:55:34.000000000","tz":0},"subject":"Drop support for python 3.9 and 3.10","message":"Drop support for python 3.9 and 3.10\n\nHibiscus (2026.2) PTI points to 3.11 as lowest version [1].\n\nMake sure relevant jobs are triggered by tox.ini changes.\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I5ee2ba55316fbad5dab6d63131950b3242df9314\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/573d4ccdd553971d268532beb807d8ee17bf25b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/573d4ccdd553971d268532beb807d8ee17bf25b1"}]},"branch":"refs/heads/master"},"23cdc1c30ea90acc80827f8f2684b97d27fb179f":{"kind":"REWORK","_number":5,"created":"2026-05-12 06:59:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/988048/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/988048/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/988048/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/988048/5"}}},"commit":{"parents":[{"commit":"6cceba34cfa630ab18bfa831eafc7d81376509e3","subject":"update constraint for os-ken to new release 4.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cceba34cfa630ab18bfa831eafc7d81376509e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-11 08:18:32.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-12 06:59:51.000000000","tz":0},"subject":"Drop support for python 3.9 and 3.10","message":"Drop support for python 3.9 and 3.10\n\nHibiscus (2026.2) PTI points to 3.11 as lowest version [1].\n\nMake sure relevant jobs are triggered by tox.ini changes.\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I5ee2ba55316fbad5dab6d63131950b3242df9314\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23cdc1c30ea90acc80827f8f2684b97d27fb179f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23cdc1c30ea90acc80827f8f2684b97d27fb179f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
