)]}'
{"id":"openstack%2Frequirements~988409","triplet_id":"openstack%2Frequirements~master~Ic04bcb6086907465d00d06380c58ac8301c403cc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic04bcb6086907465d00d06380c58ac8301c403cc","subject":"update constraint for cliff to new release 4.14.0","status":"MERGED","created":"2026-05-13 09:20:39.000000000","updated":"2026-05-14 08:31:02.000000000","submitted":"2026-05-14 08:29:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988409-new-release","meta_rev_id":"ab771171d79ecc568d1230b62e293064e9e9fd36","_number":988409,"virtual_id_number":988409,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 08:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-14 06:28:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-14 06:28:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-13 11:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 06:28:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"edab1f3d0fe734c857169f65e88eccf452c49589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-13 09:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de72f49f543f0c62340a24c1758e07c27cc35ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 11:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab91755d70ca42349a2738053a5bb6a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f2cdc4bc804e91af8bcb85bb2e97b1 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c942cafdd0414da3b8a64f693f54c821 : SUCCESS in 1m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8f65167dbfb49a580259bd723324a22 : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f454a39c3ac9494ea072a6ccae320516 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c3daa54abfb64ad9bb6760678f82e60a : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6f7eea98f0e4102ac1dc5f21d9146a0 : SUCCESS in 2m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b4f176e680c45ef940b3eba91c6f1cf : SUCCESS in 2m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ecf91d32a2434c5db68536d653ce20f9 : SUCCESS in 2m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/41c0f81f76d0406a99a2ff2dd2e007f8 : SUCCESS in 5m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/264982d10127411386df60b88a369221 : SUCCESS in 5m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/308abaf086ee4b82bc9c3947f74ff3a0 : SUCCESS in 2m 34s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e9811af3fcd142919414d967acebdcda : SUCCESS in 5m 00s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1287d064df9544e1949ee0320f70bfbf : SUCCESS in 3m 22s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/90da863e7fa5439f93f817cda89e75f3 : SUCCESS in 7m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6f05af53276d4e469d913ce17ea0cc16 : SUCCESS in 4m 10s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/014950a94c304ae69c3277cc77231723 : SUCCESS in 5m 10s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3b05d629406b443ba7a847ab7249a0dd : SUCCESS in 3m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6474ea032e3449b1ab36a262793ac108 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59952da6297840caacefc0338cd86b7e : SUCCESS in 5m 13s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b55ba3f97f0b4dfc9b63e52cb59f25af : SUCCESS in 4m 31s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0ee5bd48f4314782ac2c2f9c475f74f4 : SUCCESS in 6m 45s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9f5375b4fcee4a61b20a175398edf80a : SUCCESS in 4m 05s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/612c45359d514b99b7518e8c46537000 : SUCCESS in 11m 58s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/576fec7a9d2248d3bb83f83aa10ecd2b : SUCCESS in 2m 20s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/69adc5a75a464d98990e0d40c838f2ee : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ee04658f3cce41a79450edf512dd0074 : SUCCESS in 9m 29s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c60184900c074973b5404bcf34d8c5bb : SUCCESS in 18m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e2960fa873a43c5a3e8760d9ffc730a : SUCCESS in 6m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/061e9638c12e450988894375852064af : SUCCESS in 14m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8bb827ebd1bb4c2dbb530eca51d9411b : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/288f1777fd2b433c820b3bd9197a0596 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/661441782dde45d8ae800dd0a4537f62 : SUCCESS in 24m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e6e0f68fa7de491db4565549b0c8ba80 : SUCCESS in 3m 15s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/276dd20bd7b1497b9af43b5e9d3b72dc : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab10ec63edba4a14959cf5e3fe852704 : SUCCESS in 7m 18s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2ac7d08d87b04a588dad3268fe330220 : SUCCESS in 19m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ce407a8e11e45bd8f6c39133e6922d5 : SUCCESS in 6m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5976b52f56f04463a315a029fe6d5650 : SUCCESS in 1h 02m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9485e64e433441b5a6f4a9a942573381 : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d3b29d703ed7fb6a4d819dbe74562cf01feed9f9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-14 06:28:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61629d916c88f393c7ea96841977e8bdfac4ff69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 06:29:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c57a1ba3c1a3b2d4dbf190ec52a1ec906021582b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 08:29:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ea9eab841d44c3a01baab0e99ef048\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c450bf40154f30b11cf07d2f81a88c : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc2a68cf2bd4746883609d246794dfc : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab5b6ae14e844e77bf72cb40e60cbac1 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ea1566262be4492a449092ecd5ed8d4 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7b82ff5b48147e4b9fc9f6a69c82d06 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ceb1bdb7546a4f4d946158ea28623009 : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66317516b1a74fc3be2a72b3ca8589ab : SUCCESS in 7m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6df6b75d09be460aa73fa484585f0faa : SUCCESS in 8m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0cc2dcec820e4c3786e310776421590e : SUCCESS in 5m 19s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5ba9ac0e038640afa96b5c87eb6524e8 : SUCCESS in 10m 37s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/78804237b90c4d16afda2f8d0322e3e7 : SUCCESS in 4m 39s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/68801fe982154f2eb881eef2042ddc51 : SUCCESS in 5m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f2972a4e501c4b5a9da3d34663e8dedd : SUCCESS in 3m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c04096fe10ba4d1fbb8745c9f0d6a04a : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45fe7dd631ae4c799dccaa3a23c49fe0 : SUCCESS in 6m 30s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/77615790af1a4265824412db6b0483d7 : SUCCESS in 7m 12s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8713691206e845c99c4ec83c9a0fc7ed : SUCCESS in 10m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7c70de77b4c841f999a7425b7e731ccc : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/10b3cff2b8074621b0ef4ef5704ebe4f : SUCCESS in 8m 08s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d9624a8bac5c454ca0c51325bd2ec999 : SUCCESS in 28m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b2d7c76b997a41289af3dd76e0cd3adb : SUCCESS in 7m 00s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c9053a54adc247f6ac8399eaa4f01492 : SUCCESS in 13m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/aabb47fafa3f40e1a1e92d445bcd8088 : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c758e9ccdc4145868ae1a2547556d637 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f58b0eb4bf204b06a16a4f276481e8c4 : SUCCESS in 25m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/424f71b86944474d9fe923d8b3f17635 : SUCCESS in 5m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/036341fc05b44a6cbd1b9bdf991eade9 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2767a7dd38dd4ef2bc7e879180d94c90 : SUCCESS in 7m 19s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/528f04d0374b42cea433c56fed988777 : SUCCESS in 19m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2825d810c0c94cbe9fb33ed3095f76fb : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15e119ab30d643e9ae69bdce855ad398 : SUCCESS in 1h 01m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd4abc691bd940db9207a0dc33fd3046 : SUCCESS in 1h 54m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d844f627b5bb7f601d05a26b86ea3cc40bf5ff88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 08:29:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab771171d79ecc568d1230b62e293064e9e9fd36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 08:31:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aa35c2ebc2c4907a8139db071e201ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782dc66ffa824e02b7a485aa43b8cfff : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c8c70527d6053b8723cb332b61ea74e57dda634","revisions":{"6c8c70527d6053b8723cb332b61ea74e57dda634":{"kind":"REWORK","_number":1,"created":"2026-05-13 09:20:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/988409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/988409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/988409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/988409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/988409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/988409/1"}}},"commit":{"parents":[{"commit":"4cd3abcd0aa244835cd532d2c1fbbebfd48e5c26","subject":"Drop py310 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4cd3abcd0aa244835cd532d2c1fbbebfd48e5c26"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-13 09:10:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-13 09:10:37.000000000","tz":0},"subject":"update constraint for cliff to new release 4.14.0","message":"update constraint for cliff to new release 4.14.0\n\nmeta: version: 4.14.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2ae2a812f3ca9d3f846129e9853a69c01e159fa4\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ic04bcb6086907465d00d06380c58ac8301c403cc\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c8c70527d6053b8723cb332b61ea74e57dda634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c8c70527d6053b8723cb332b61ea74e57dda634"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
