)]}'
{"id":"openstack%2Frequirements~988413","triplet_id":"openstack%2Frequirements~master~I0aae2a15831dc7c078cf34ca84cd0277ab4be87f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0aae2a15831dc7c078cf34ca84cd0277ab4be87f","subject":"update constraint for python-zaqarclient to new release 4.5.0","status":"MERGED","created":"2026-05-13 09:37:38.000000000","updated":"2026-05-14 09:22:53.000000000","submitted":"2026-05-14 09:21:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988413-new-release","meta_rev_id":"de37f115e81f0e585f1feffaa2338fa96cc7029e","_number":988413,"virtual_id_number":988413,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 09:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-14 06:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-14 06:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-13 11:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 06:48:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"10cdbb068012e18f6026adc573252ef2a41b56ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-13 09:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40d455c037db903c0118db6f114591d99a57bc1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 11:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd5d700e732456d91eb9aefb9180b7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd95b37ddf04c738853a5f526670803 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0595b8e64c374819870f2d8125de4fcb : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf1e886859d0493eb31bea94a4b6f87f : SUCCESS in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4daf351a4d104a7cba4efe12eeda4494 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/130c02a4f59249499964e1865cf2e7f6 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fbb4aa0e1e741c3b5c36084ee26b6e0 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/029693348f8e4c5a804c591d97fed5b6 : SUCCESS in 3m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/62a79d9c3114420e9e85fda18c6b9c77 : SUCCESS in 5m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dd9209b3c48f40308bd82dd451716baa : SUCCESS in 6m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6c066ae79f44ecb95a010d2a8aa2b88 : SUCCESS in 5m 11s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a87d963b788a48fb8e2efa5aa56a9880 : SUCCESS in 5m 59s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8091091c0218410bbac8bb2ebf471ead : SUCCESS in 5m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/47fdb19abb724d3ea56519e4e09ebb36 : SUCCESS in 5m 29s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d2c43b93021f4520a8fa3f522073b750 : SUCCESS in 11m 08s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fafa524e49234862a7adf8ba04c32a3d : SUCCESS in 5m 03s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5df32e170b1a4b92874a64a8498b6a67 : SUCCESS in 6m 54s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ee940b2f53b242409f3d4470c7d9b354 : SUCCESS in 6m 58s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/32c04fec81734743b2c8594d38106d36 : SUCCESS in 14m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5e4f6c7a9ca4985844d1de2bec17d4e : SUCCESS in 6m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/56901fd9ab1f4ec291a84ed45df33185 : SUCCESS in 7m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f43699ae5c7045229ff1092f1df11477 : SUCCESS in 11m 02s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9922f20d3466455b89a168cc927762cb : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/599089865be049e6a0242dbb16249aab : SUCCESS in 8m 59s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/663751d49a2d4879a24cfad1dfdf3beb : SUCCESS in 3m 39s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/86cef1d1733548c1904b2e4415b172f7 : SUCCESS in 4m 58s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0ad6ffb5c25e4a2db7c333c552a95b01 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f8e4b8bd51974d038e982fc64c21b9b8 : SUCCESS in 28m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1e876a7ee2414027884b65e4afe0b429 : SUCCESS in 4m 51s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/bf71a310eb3e430592f932e54a8c1423 : SUCCESS in 12m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8cb1d958f07f48ea94d12092e3747c6f : SUCCESS in 3m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb5fd8cd450746739b92b248d52713c6 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2b3b50554f54fca88713823a5953a1d : SUCCESS in 23m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/31631b458bbd4ac6b631e24db0909b4c : SUCCESS in 7m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b1e1f51feb3147099eb8f98c7bd5b3f0 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1f45cb50eb94a849fd236c4d8e8ca51 : SUCCESS in 4m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2a76c1073f0b4883bcbc7273bf6149d7 : SUCCESS in 7m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/30ac331d7a3e41bdaec8729fa478a0bc : SUCCESS in 5m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6df6991e2acc4e539512aa2fb1e06228 : SUCCESS in 58m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b936e0c578754b28ae516b7904034817 : SUCCESS in 1h 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"53f1d68fbc8a1f69a3c1ba0e394465b7552fcd5e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-14 06:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4819b1fcd1114741d7baca0ccbad727f76ff4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 06:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e053026bf247c7b224fed3a8715cb30ef1be95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 09:21:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d7f8d7702f407f86620518063c2ee5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a33b5ee2b754fa69014abccc97ff358 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc177a18414c4b3c88cb0c3a7457a688 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b85614c6044249ffadea3466f4b19318 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ca9311433c6435fb65dd76eed3443c3 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/743b3dc02ba244ef894e609cf1491c1a : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6052af9fecf403e82dac05463a67af2 : SUCCESS in 2m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/746c2439fb6247189daf6230a600a48e : SUCCESS in 5m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/83a28ed82bf446b38fdc01dbaae1e52a : SUCCESS in 9m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1d3130676f2c4a729962971de640af10 : SUCCESS in 7m 09s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/49d4892aeb554888a042b8f78f9c569d : SUCCESS in 8m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c67f53124150486d9ed2624bf1dd9eab : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1033a950308a4830bb3ec100d9e35934 : SUCCESS in 3m 29s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/867a9b8d261f414ebfa6e26a93b65358 : SUCCESS in 7m 59s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b82aab8948104dac80072678e54ffba9 : SUCCESS in 15m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a180bd64ecfd40b78a30ab2810b4d64d : SUCCESS in 5m 59s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8ffc393730cf488d882706fdc7ad579e : SUCCESS in 7m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f37cfe5eabc949668854e1e97f9d64f4 : SUCCESS in 10m 51s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5e874b5a3b924ed59573ebd21f379893 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6e67663e197d4ff0ba5c33fbbfd880d9 : SUCCESS in 8m 21s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dfe355b91ae34fb28d91b1dbebacb186 : SUCCESS in 18m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/403564e8a1b74f189695d28daf65bcb7 : SUCCESS in 7m 59s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2fa78695423f4ffa99bf624bf8288b46 : SUCCESS in 13m 33s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b8285b6a610942c09e6dbc6cf0c4833e : SUCCESS in 3m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cc3959f82c34ff79f42533e1907dac0 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/910dd5f206c24f679245db7f927fa12a : SUCCESS in 29m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b6210a7314e04b47a20ff29b526e12ac : SUCCESS in 5m 46s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e1bc6e4f44a0427892905952c03b24ae : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/286eebd6043f4ce9bd2aec31f2c67929 : SUCCESS in 6m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/87a7a588036048a9a4002afa24fdb85b : SUCCESS in 18m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/984fb9a6d527429d9e800a4955d48335 : SUCCESS in 2m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddf2220ec2b24e55b9a73dd12122d50a : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e69a92f707b4def8cb50e5a4089c2c2 : SUCCESS in 1h 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f861e8559a435dccde90a496300eaffdb56857a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 09:21:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de37f115e81f0e585f1feffaa2338fa96cc7029e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 09:22:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca3ec7745e3740b888a5f44dd5c15282\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a755411f9bfc4380b94f94c39c23a642 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d0b9c386738d983f1ec694d0502cae118594d0e","revisions":{"1d0b9c386738d983f1ec694d0502cae118594d0e":{"kind":"REWORK","_number":1,"created":"2026-05-13 09:37:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/988413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/988413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/988413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/988413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/988413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/988413/1"}}},"commit":{"parents":[{"commit":"4cd3abcd0aa244835cd532d2c1fbbebfd48e5c26","subject":"Drop py310 job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4cd3abcd0aa244835cd532d2c1fbbebfd48e5c26"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-13 09:27:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-13 09:27:36.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 4.5.0","message":"update constraint for python-zaqarclient to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: team: zaqar\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I93238d267c072d0723feaef36eea6a85f2da4f0a\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0aae2a15831dc7c078cf34ca84cd0277ab4be87f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d0b9c386738d983f1ec694d0502cae118594d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d0b9c386738d983f1ec694d0502cae118594d0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
